)]}'
{"id":"openstack%2Ftrove~759317","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~Id85c25c0cd3903f1b536a08ddcf69e306ce7345d","project":"openstack/trove","branch":"stable/victoria","topic":"user-port-stable/victoria","hashtags":[],"change_id":"Id85c25c0cd3903f1b536a08ddcf69e306ce7345d","subject":"Fix getting user port for instance","status":"MERGED","created":"2020-10-22 19:34:39.000000000","updated":"2020-10-23 01:29:41.000000000","submitted":"2020-10-23 01:28:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759317-1603416500908-7901bf70","meta_rev_id":"b6d54bd10f23bd40bb459b6853c6be58dcbe15f6","_number":759317,"virtual_id_number":759317,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-23 01:28:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-10-22 22:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-10-22 22:56:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-23 01:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d200250eb4b195b4cf04cea52e2a9e2b4c539ef4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-22 19:34:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0a22ea16c9a89eea15e702ad14d38417d9dd8fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 21:48:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd83bb0cf9a94107a70e9245ff4a0e59 : FAILURE in 8m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43aad73b741b486ea3067cf41a355c18 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c2758d6d77e4b478ff4eef43b662213 : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4143ff74e3714736825a8f9e0147fa22 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68a530bc8932469a94e6f1897fa8bacb : SUCCESS in 17m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d63d6e3af28b422d86c6eba9c1160425 : SUCCESS in 7m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/76fddd372de547f5b1186c7aa8f308c0 : SUCCESS in 6m 54s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/79b74012e8d44c4e93f7530d13ea9636 : FAILURE in 2h 09m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8973637395ef42c8bc241088ac1ec405 : SUCCESS in 1h 05m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43fe0a483eb0a7aa700dade4dd2d0d7eff503771","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-22 22:56:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"55410958abaf0d85d4d35762e4f9baf1d66f6434","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 22:56:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fe4b33c37814e9e4907850e8dcfb185e2758264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 01:28:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b101f19d8df468faca723924532afe5 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9973fa7de4484fb4f6ead308c5dbda : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6561155cba7045e4904482b97f4c0240 : SUCCESS in 19m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d659613e86e948e4bb271b10f1da603d : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a71d6a916b4e49b8fb254633982899 : SUCCESS in 9m 05s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a21aa19bc1ff44cdb772f7841b248ed0 : TIMED_OUT in 2h 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3f032d588325d762ff28bfee1d5d6077bbd88eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 01:28:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ed3b6aed85d2ba3d4472125f378a13eb6196aaea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 01:29:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cdf0e63e4f54038af72b95af3931e74 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"666b83aeab13af3dd54895a25ffd8816a51c888e","revisions":{"666b83aeab13af3dd54895a25ffd8816a51c888e":{"kind":"REWORK","_number":1,"created":"2020-10-22 19:34:39.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/17/759317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/17/759317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/759317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/759317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/17/759317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/17/759317/1"}}},"commit":{"parents":[{"commit":"6e1af9719210f8e5e6e3ba90bea15fde3cd8b777","subject":"Merge \"[TrivialFix] Fix \u0027--replica-of\u0027 argument format in docs\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e1af9719210f8e5e6e3ba90bea15fde3cd8b777"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-22 10:12:50.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-22 19:34:39.000000000","tz":0},"subject":"Fix getting user port for instance","message":"Fix getting user port for instance\n\nCheck the port type by network rather than port description.\n\nChange-Id: Id85c25c0cd3903f1b536a08ddcf69e306ce7345d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/666b83aeab13af3dd54895a25ffd8816a51c888e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/666b83aeab13af3dd54895a25ffd8816a51c888e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
