)]}'
{"id":"openstack%2Ftrove~759647","triplet_id":"openstack%2Ftrove~master~Ib885a05eaa74fc1566a398935402b76ca414e99e","project":"openstack/trove","branch":"master","topic":"mysql-8-release-note","hashtags":[],"change_id":"Ib885a05eaa74fc1566a398935402b76ca414e99e","subject":"Release note for mysql 8 support","status":"MERGED","created":"2020-10-26 07:56:30.000000000","updated":"2020-10-26 23:16:13.000000000","submitted":"2020-10-26 23:14:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759647-1603754099106-ed70fdfb","meta_rev_id":"23371ace17db1ac7b857af96a51f385fba358c5f","_number":759647,"virtual_id_number":759647,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-26 23:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-10-26 23:04:15.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-26 09:52:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-10-26 23:04:15.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-26 09:52:24.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-10-26 23:14:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ef2a1269b31d5161cbbf0f80fbec091ab5878d2","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-26 07:56:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c115eeac1e09eaf95d43b11d6503ecf9f51c51ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 08:15:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7133e5bde57e40a48039450458def2a0 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/047488269e4c4f65be3f27f168864e9d : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc4b61ebe6de4f5988b10e3bf6f21fb3 : SUCCESS in 13m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8ecbd14afbd4e41797b431a06953e2728c4d216b","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-10-26 09:52:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"862243954817261467e6f0712533989518001acb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-26 23:04:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aa96d74fe15f98a6151f176069b689a3b0b5b9a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 23:04:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c63b112a62d940eb45dd70baed3d539c31bcd7cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 23:14:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc6aa000807645d18a1ab2e22f80eb0e : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bff8fbee28246e2bf766ac0ab576f77 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f227e55da620499ba2bcaddaf551e482 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ad89c5a04eb10c33c3e8f33a139dbe6bc3f9b08a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 23:14:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b333b1004471ab7ff22d8d874b06f21a101cb61e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 23:16:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b7abf1b84d1402c8b272cce77a9060b : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/446da5adb5904196b15f3240ead2e57d : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33f5c2d63d407b4ca0799f2502eb165434f31aee","revisions":{"33f5c2d63d407b4ca0799f2502eb165434f31aee":{"kind":"REWORK","_number":1,"created":"2020-10-26 07:56:30.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/47/759647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/47/759647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/759647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/759647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/47/759647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/47/759647/1"}}},"commit":{"parents":[{"commit":"d1af33f17b0994ac1d0ca5acca91f2f29bc82ce9","subject":"Support mysql 8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d1af33f17b0994ac1d0ca5acca91f2f29bc82ce9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-26 07:56:19.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-26 07:56:19.000000000","tz":780},"subject":"Release note for mysql 8 support","message":"Release note for mysql 8 support\n\nChange-Id: Ib885a05eaa74fc1566a398935402b76ca414e99e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33f5c2d63d407b4ca0799f2502eb165434f31aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/33f5c2d63d407b4ca0799f2502eb165434f31aee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
