)]}'
{"id":"openstack%2Ftrove~760257","triplet_id":"openstack%2Ftrove~master~I558c36b4f0894f8476ce7fc9c88c1829e8762011","project":"openstack/trove","branch":"master","topic":"list-order","hashtags":[],"change_id":"I558c36b4f0894f8476ce7fc9c88c1829e8762011","subject":"Return instances by created order","status":"MERGED","created":"2020-10-28 23:25:44.000000000","updated":"2020-10-29 12:04:32.000000000","submitted":"2020-10-29 12:03:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760257-1603972992522-05be731c","meta_rev_id":"69eb7a83df120836d66d234e7438dd95668d5dc1","_number":760257,"virtual_id_number":760257,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-29 12:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"recommended":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"all":[{"value":1,"date":"2020-10-29 03:11:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":2,"date":"2020-10-29 09:43:06.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-29 03:21:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2020-10-29 09:43:06.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-29 03:11:56.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-10-29 03:21:12.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-10-29 12:03:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47a3b74d531d74173d88561c062af17dd60ff098","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-28 23:25:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"427f6e89a199662195c9d0ec2281e88f541d95ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 01:46:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad196433d14a4b2b9312e18bcf3706a7 : FAILURE in 6m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/603efbcc10694b77bc5790bce98d1894 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad1dae65cec4ca48fa41389f867abef : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e6c8b4fd9934c30be360d42a63ac230 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e63a4dffee9f493f82728c4b6745665f : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b3863896484b268beed445d8d3ecf2 : SUCCESS in 5m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fc3e628cb01246e0bfabc41d6e8451c8 : SUCCESS in 5m 23s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b6d6460a025143b38176f01163eef857 : FAILURE in 2h 13m 36s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b8e80a2b7bb4b58a6b043a1ac1eb2ac : FAILURE in 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3582bfe2b29c513f329e241806c409ec0bc053d","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-29 03:11:56.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good, would it not be better to do this at the DB layer somehow though?","accounts_in_message":[],"_revision_number":1},{"id":"0cecc5fd0bb50df22dba41ec06170f0b334952d1","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-10-29 03:21:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"03e23bb09a13da6d8647e01e6be2be4e169debf5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-29 09:43:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n\u003e Looks good, would it not be better to do this at the DB layer\n \u003e somehow though?\n\nThanks for the review. That\u0027s part of trove technical debt, implementing in db layer is optimal but need some refactoring.","accounts_in_message":[],"_revision_number":1},{"id":"42d64af73e3270bf6a79f31ccae9bc13507d34a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 09:43:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4615c323e0d73c3191c6ec3193a7fb00ce36cb32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 12:03:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/064245852fa64843a84a17211b0c3def : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2cc6c1240248b9bbbf178e25d235d2 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52f813fd169b433592627b690eeaf99a : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8dc0fc940e54d5cb277ff83c4db2cf7 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365bae73060f4395a0d3badac991228c : SUCCESS in 5m 44s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/eb8201e624214823b6d607ae51b58c73 : TIMED_OUT in 2h 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"714a5cb520fe1f93f2a60e64ac0809b7aacf2631","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 12:03:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"42f95370117ea31c26abc9602ba511fb14facc49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 12:04:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e31a4de7d004f058ec8584426ad3a02 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69c8467854795fd3a01d3fd0865aceb2c1806219","revisions":{"69c8467854795fd3a01d3fd0865aceb2c1806219":{"kind":"REWORK","_number":1,"created":"2020-10-28 23:25:44.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/57/760257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/760257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/760257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/760257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/760257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/760257/1"}}},"commit":{"parents":[{"commit":"d21d37d6285fa949f286c86ccbccd8fc869bb5bf","subject":"Merge \"Release note for mysql 8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d21d37d6285fa949f286c86ccbccd8fc869bb5bf"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-28 23:09:14.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-28 23:25:20.000000000","tz":780},"subject":"Return instances by created order","message":"Return instances by created order\n\nChange-Id: I558c36b4f0894f8476ce7fc9c88c1829e8762011\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69c8467854795fd3a01d3fd0865aceb2c1806219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69c8467854795fd3a01d3fd0865aceb2c1806219"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
