)]}'
{"id":"openstack%2Ftrove~760260","triplet_id":"openstack%2Ftrove~master~I7abf5563b22ab1914fa355c089a3793da07f2215","project":"openstack/trove","branch":"master","topic":"backup-decryption","hashtags":[],"change_id":"I7abf5563b22ab1914fa355c089a3793da07f2215","subject":"Fix restore encrypted backup","status":"MERGED","created":"2020-10-28 23:41:43.000000000","updated":"2020-10-30 07:26:56.000000000","submitted":"2020-10-30 07:25:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760260-1604042721716-3f776333","meta_rev_id":"b913855c7febb1d6ab4be859143deb335c645a25","_number":760260,"virtual_id_number":760260,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-30 07:25:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-30 05:41:55.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-30 05:41:55.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 07:25:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4824068587e703eb66a57f4eab919c1e5614f497","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-28 23:41:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05312f144ce4a667dd00b77fa15b9b33f191d04c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 02:34:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9265863c71d5477597bcef5b8cf45ffc : FAILURE in 8m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01d477a629a344bf932bb4b2a3b69148 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c42e4c96e294475d8dc545960dc5c892 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/572e67d29ded459ca3b609198a3679c8 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66ecc67acecd4359b132094f3a23fc0f : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3003a82d2d004e109312362bd2ebd491 : SUCCESS in 7m 33s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6cf969979f824c228455ae16befe9b06 : SUCCESS in 8m 05s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a540c8d563e84406a591d07fe12d4ffe : FAILURE in 2h 09m 18s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0eefced9f974718945c5cac81d62d4c : SUCCESS in 1h 04m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"872f67a1c30ca7e6cb3eb437a4b283373b9ee1ba","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-29 10:06:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"90ee3bf5960b72d749f50842401d2bbb5cbb347d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 12:07:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c13709c4598849bb9d0f2e2f0813601c : FAILURE in 12m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be6b8af45f284f938c86000dc5e3d481 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5115033e7774018ac2dd3074814ba32 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39c6ceeb145648e9a17836e485a2a119 : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9262096279944c2199a492a19c7c589d : SUCCESS in 12m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b740ecb559ad4f6c82f0aaa809fec52e : SUCCESS in 10m 51s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/380191d73aff45ec831e405577f35bfb : SUCCESS in 6m 29s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e7a2778c46374f1aaf518ad044a800e1 : FAILURE in 1h 52m 29s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61f2328aba87454b9cae3620ad1be5d8 : SUCCESS in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c97d85e12ed5e90fcdd3bdb7a3a65d8f21d894c1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-29 21:43:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a0c01fce2df9dd0f04bcfa38b77527f14930374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 00:06:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ceda52fff4e7457fb7d707a917a9f84d : FAILURE in 6m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54f86f325add471ebde07fc7e9e3462a : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a2ef44cd70b4952a094a338ea466c46 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6bce637c7f7429fb84bced2f7221a8a : SUCCESS in 13m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fe675e6dd4a473e827b957623855752 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61e8d8e48d5140faa63d5954e2a57fbc : SUCCESS in 7m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8ef3e711b1f34d408c7b2c6aef546311 : SUCCESS in 5m 43s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bf3d7977ce35467d913bd2518dbc9a71 : TIMED_OUT in 2h 13m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/678aaa172b354681b100b794bfdc1ec7 : SUCCESS in 1h 03m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"80b9f735d724a444468e307b9cefa6a764eaf2f0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-30 05:41:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f3469e905cdec1115389cb4c5c22dfa9a0a89770","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 05:42:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c358d9e22a6f9578f4d9f7688038b6786a12e007","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 07:25:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8c2f74f112d4204b089e7107e51bce66763100b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 07:25:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a23db08e095f4be9b827f703d48d3f82 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d565b1621234444fb77c74b5076d32e2 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f2bbba6e0be4430a00349c1862b304b : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf01c651128a48db9c4a96a71271cbc2 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d134bc7bee494a8abf920f692ad7f631 : SUCCESS in 5m 21s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/52ff09e8cc99421d928de381639bcd0f : FAILURE in 1h 41m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c83719cdfda7d0eae84d8f638d0f1aeeefe3c77b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 07:26:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f36c2b74f2f44d9a9d72c47d6228ca0 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bd2b256a377a5648586ec4a5d51edd58dd88a620","revisions":{"e36bd823fb69d066dfebce2856af3608efac756c":{"kind":"REWORK","_number":1,"created":"2020-10-28 23:41:43.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/60/760260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/60/760260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/760260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/760260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/760260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/60/760260/1"}}},"commit":{"parents":[{"commit":"69c8467854795fd3a01d3fd0865aceb2c1806219","subject":"Return instances by created order","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69c8467854795fd3a01d3fd0865aceb2c1806219"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-28 21:55:22.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-28 23:41:16.000000000","tz":780},"subject":"Fix restore encrypted backup","message":"Fix restore encrypted backup\n\nFor the backups created prior to Victoria which may be encrypted, the restore\nfunction in Victoria or later release should correctly decrypt the backup data.\n\nBackup encryption from Victoria is disabled.\n\nBump the backup docker image tag to 1.1.0.\n\nChange-Id: I7abf5563b22ab1914fa355c089a3793da07f2215\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e36bd823fb69d066dfebce2856af3608efac756c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e36bd823fb69d066dfebce2856af3608efac756c"}]},"branch":"refs/heads/master"},"68b6784d6857c810d7885aec16e166d418789e20":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-10-29 10:06:59.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/60/760260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/60/760260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/760260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/760260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/760260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/60/760260/2"}}},"commit":{"parents":[{"commit":"69c8467854795fd3a01d3fd0865aceb2c1806219","subject":"Return instances by created order","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69c8467854795fd3a01d3fd0865aceb2c1806219"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-28 21:55:22.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-29 10:06:52.000000000","tz":0},"subject":"[WIP] Fix restore encrypted backup","message":"[WIP] Fix restore encrypted backup\n\nFor the backups created prior to Victoria which may be encrypted, the restore\nfunction in Victoria or later release should correctly decrypt the backup data.\n\nBackup encryption from Victoria is disabled.\n\nBump the backup docker image tag to 1.1.0.\n\nChange-Id: I7abf5563b22ab1914fa355c089a3793da07f2215\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/68b6784d6857c810d7885aec16e166d418789e20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/68b6784d6857c810d7885aec16e166d418789e20"}]},"branch":"refs/heads/master"},"bd2b256a377a5648586ec4a5d51edd58dd88a620":{"kind":"REWORK","_number":3,"created":"2020-10-29 21:43:49.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/60/760260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/60/760260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/760260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/760260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/60/760260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/60/760260/3"}}},"commit":{"parents":[{"commit":"69c8467854795fd3a01d3fd0865aceb2c1806219","subject":"Return instances by created order","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69c8467854795fd3a01d3fd0865aceb2c1806219"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-28 21:55:22.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-10-29 10:19:03.000000000","tz":780},"subject":"Fix restore encrypted backup","message":"Fix restore encrypted backup\n\nFor the backups created prior to Victoria which may be encrypted, the restore\nfunction in Victoria or later release should correctly decrypt the backup data.\n\nBackup encryption from Victoria is disabled.\n\nBump the backup docker image tag to 1.1.0.\n\nChange-Id: I7abf5563b22ab1914fa355c089a3793da07f2215\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd2b256a377a5648586ec4a5d51edd58dd88a620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bd2b256a377a5648586ec4a5d51edd58dd88a620"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
