)]}'
{"id":"openstack%2Ftrove~760284","triplet_id":"openstack%2Ftrove~master~I455477a2e7a00f0d132971a2a684352967ac19b9","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I455477a2e7a00f0d132971a2a684352967ac19b9","subject":"Support ram quota","status":"MERGED","created":"2020-10-29 05:11:23.000000000","updated":"2020-11-11 14:29:52.000000000","submitted":"2020-11-11 14:28:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760284-1605104889496-28503777","meta_rev_id":"bdf94a46538f741b2b53792c7830ef89275d3740","_number":760284,"virtual_id_number":760284,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-11 14:28:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-11 09:53:40.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-11 09:53:40.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-11 09:53:40.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-11-11 14:28:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd3f98486315ace0cef9cf97f8853d47e3ca7092","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-29 05:11:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"623bbfee346ea8dafd5301dd3a64a8b600165b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 06:59:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cef66039b674d13be819088f6aaa82a : FAILURE in 6m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c247077e4ad424ba60af8964888da3f : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb3fb81a4b649128b0a62c61f60bee2 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7586a0c9c96948cbb7cfd98d4c414328 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a106f5c1d15c4487816710036fbf9125 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1cbe3cdcce412e91aaf88e95844a85 : SUCCESS in 5m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b3262dfc4daf4ee8bde7e565160bfaa0 : SUCCESS in 5m 43s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f55768bdcc534d1ca27a210bc219b1a0 : FAILURE in 1h 42m 18s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa906d823a12487db2198d05e6579f37 : FAILURE in 1h 09m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"752551dda09aee1e5a7f54c71cca98573a8e4936","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-30 04:29:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc79be47ba6dbd11b45e9cedeb787721e8d84197","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-30 04:43:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ff5da857438328ef956616f48e6007581dbe85a0","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-30 04:43:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f7353c3fba08172ceaa2b70f8dfc2c1e00d615d8","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-30 04:54:23.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"8e57ea3c6cf10c6088e211c14b6d078043402348","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-30 04:55:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"21f958e69ecb8ce3213d334f6e982bf96608409e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 07:17:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d07ed0f5a5645189907f621aea02338 : FAILURE in 9m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63ebf2efc47f4b319551c584defc2870 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd72ae7ed18846e7bfb98d5583236b13 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86371d751fe24d349e1b4b298b5f1208 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4374d500ce54f44bfbb0e764b29b640 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32a89d0760524dc182246e37ac60da40 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ecc50509f4f4dbfa634cdeb75069e3e : SUCCESS in 5m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/53c18c95decd48e6865758d644c19bad : SUCCESS in 5m 30s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1f10f5bcdd1a479783228067863a6964 : TIMED_OUT in 2h 13m 18s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cbec5fd767049f18a4bfd3af41c4968 : SUCCESS in 1h 00m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"35aec9e12bdfece3911436cc3ae4afd902302b60","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-31 08:22:40.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\n1. Please update API doc.\n2. Change here https://docs.openstack.org/trove/latest/admin/run_trove_in_production.html#quota-management as well.","accounts_in_message":[],"_revision_number":6},{"id":"09093bdcbf69b0904041892360c194f08444707a","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-03 20:58:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f2c200077284dac163f8da8061c4f23487a863a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 23:13:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a494cb13da34344b9dcba70c49e6ba5 : FAILURE in 6m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04f375e33b364d179209775548567ac4 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee807850ee254cea933f85d08d13c078 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bba82ebab0f4dd2a32529e485a1ab0c : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b294b12714e041ca96d3527ca4f91a00 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db7a9ef3dda142a986ccb4df7ab706ef : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7369addc534b5981949029cdf77e7c : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47559a3e758444bd87343fe9f2c3e936 : SUCCESS in 4m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/542bcdf5a33544df94d0b052213c8d29 : SUCCESS in 5m 46s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5ed4a2485bff41db90f10782fd45b49d : FAILURE in 2h 04m 34s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66c1f9a9a3544013a2e70261c4e9d9f5 : SUCCESS in 1h 03m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"44f45e0b00cf2e762512378d61c602521fe70eb5","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-04 03:55:45.000000000","message":"Patch Set 7:\n\nHi Sam, is this patch ready for review? Have you run functional test on your dev env?","accounts_in_message":[],"_revision_number":7},{"id":"673bd73e3cc985c3a13e3f5ad10cb81b1f568fa2","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-06 04:41:42.000000000","message":"Patch Set 7:\n\n\u003e Hi Sam, is this patch ready for review? Have you run functional\n \u003e test on your dev env?\n\nYes ready for review, works in my env but I would feel better about it if it was tested in a different env too","accounts_in_message":[],"_revision_number":7},{"id":"97e0ac3b680eeac179caba107ce12fe4538e0b83","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-09 22:43:24.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fe2ddeaf766835c8f77b0427ada520733a7dfcf4","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-10 23:05:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e53a48202ff50fb2f3637f3b03f5c0c7b12d8691","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-10 23:06:44.000000000","message":"Patch Set 8:\n\nTo migrate I will prob just do a sql query as currently all our DB instances use the one flavor so it is easy","accounts_in_message":[],"_revision_number":8},{"id":"a1781a3a813bf8ebdcc2613b08c5eeef0ef582df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 01:27:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5f1771fd4a94e61813b574955960560 : FAILURE in 8m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd8380e680304e239759b11702f35f79 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebae08b862c54eaf9017e1c22d4d21f2 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bbbcc6b30f243af9ac04c2da410cb1d : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/754d5bef77f841c6ae72cf3d6bc01c42 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8bf67a56d2e402a97eb141f85b014d5 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32508176eb254c4492ce75d8c7c7b18b : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2128d50379484102ac79609e6d85f8ee : SUCCESS in 5m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7f2bb9adc2e9400b852bba1911c10e2e : SUCCESS in 5m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a9244007d0cf4200ad8f41f963c00c69 : TIMED_OUT in 2h 12m 40s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eba06f7fce4a4f0283cdca5ca5dee3f2 : FAILURE in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5f786f757e95c4c012b5db8ab38eed41e9eaf942","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-11 09:51:35.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"50be555cbb1360621dfeb85b3e3c447b46923d9f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-11 09:53:40.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n\u003e To migrate I will prob just do a sql query as currently all our DB\n \u003e instances use the one flavor so it is easy\n\nok, sounds easy for you. Well, we are not going to use this quota, so let\u0027s merge this patch for now. Maybe in the future if someone ask, we need to provide a script for the calculation.","accounts_in_message":[],"_revision_number":9},{"id":"3984bebb35a573e11c93b980a7057e114ca328c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 12:08:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7acb52f0d06a45e68a58a10822ef1940 : FAILURE in 7m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c0c1052f89446d5b1faa91c0c0be125 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f29d40d05244790b2ea17117ae536fb : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a471d95ece944b5845404fc69095e81 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ba1d9b71611418387cfe2e8e9f8f66e : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ad2697389944acda0d6f826d37b19a9 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/366da182fef1475882703c5526184269 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51e5e9a8108842549fcb1c4206a42d34 : SUCCESS in 5m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ffa4778fc0c346ea897bd10d17285110 : SUCCESS in 6m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5fa73b1836be46e4a0315933f2f1b641 : FAILURE in 2h 04m 26s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c82c66e25b824a1ab4057c72c961e588 : SUCCESS in 1h 04m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0fb86588c0b411760808f7e64d8731ed350585ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 12:12:14.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"697346fdef97306af13c5920733e8648e2094fea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 14:28:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"442c51e6c450848189a121f18a1202af0bc479c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 14:28:09.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f29e26257d8f4fa8956deedb8270e003 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e4afbf21b214396ae1ac98c15bcaefe : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cde7d9a25464522889998d037d39920 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/471eea0621574e44b7379ca36129ac1c : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a0ce60ac88d496c94786364bafb707a : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fabc977d22847f199bd6a9815d4f39e : SUCCESS in 6m 50s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/79199e4f20bb4163959c49374c95c835 : FAILURE in 2h 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bdf94a46538f741b2b53792c7830ef89275d3740","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-11 14:29:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f96a018b6a945ca82cbfe58b80a9ac9 : SUCCESS in 1m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f225788632124711b494b60b39a3bcbf : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"5be23d1b20bf96f9605b4ae0cf6d7eb1d17d7a2b","revisions":{"8b3e179a2ed2d589dcabbffe57858f146629fe8f":{"kind":"REWORK","_number":1,"created":"2020-10-29 05:11:23.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/84/760284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/760284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/760284/1"}}},"commit":{"parents":[{"commit":"d21d37d6285fa949f286c86ccbccd8fc869bb5bf","subject":"Merge \"Release note for mysql 8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d21d37d6285fa949f286c86ccbccd8fc869bb5bf"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 04:32:28.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 05:11:02.000000000","tz":660},"subject":"[WIP] Support ram quota","message":"[WIP] Support ram quota\n\nStory: 2008293\nTask: 41172\n\nChange-Id: I455477a2e7a00f0d132971a2a684352967ac19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8b3e179a2ed2d589dcabbffe57858f146629fe8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8b3e179a2ed2d589dcabbffe57858f146629fe8f"}]},"branch":"refs/heads/master"},"4ef3da3629ce3418b26cbbcdff1004a4a18921ce":{"kind":"REWORK","_number":2,"created":"2020-10-30 04:29:03.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/84/760284/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/760284/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/760284/2"}}},"commit":{"parents":[{"commit":"d21d37d6285fa949f286c86ccbccd8fc869bb5bf","subject":"Merge \"Release note for mysql 8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d21d37d6285fa949f286c86ccbccd8fc869bb5bf"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 04:32:28.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-30 04:28:38.000000000","tz":660},"subject":"Support ram quota","message":"Support ram quota\n\nSet default to -1 so this is backwards compatible. Existing\ninstallations will need to manully backfill quote usage for this\nto work as expected.\n\nStory: 2008293\nTask: 41172\n\nChange-Id: I455477a2e7a00f0d132971a2a684352967ac19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4ef3da3629ce3418b26cbbcdff1004a4a18921ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4ef3da3629ce3418b26cbbcdff1004a4a18921ce"}]},"branch":"refs/heads/master"},"dcb314de599eb0b7ea499eb60eea9471693db984":{"kind":"REWORK","_number":3,"created":"2020-10-30 04:43:02.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/84/760284/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/760284/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/760284/3"}}},"commit":{"parents":[{"commit":"d21d37d6285fa949f286c86ccbccd8fc869bb5bf","subject":"Merge \"Release note for mysql 8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d21d37d6285fa949f286c86ccbccd8fc869bb5bf"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 04:32:28.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-30 04:42:16.000000000","tz":660},"subject":"Support ram quota","message":"Support ram quota\n\nSet default to -1 so this is backwards compatible. Existing\ninstallations will need to manully backfill quote usage for this\nto work as expected.\n\nStory: 2008293\nTask: 41172\n\nChange-Id: I455477a2e7a00f0d132971a2a684352967ac19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dcb314de599eb0b7ea499eb60eea9471693db984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dcb314de599eb0b7ea499eb60eea9471693db984"}]},"branch":"refs/heads/master"},"704b40cdcedca36b878dfd03a2fe605aa9cf33d3":{"kind":"REWORK","_number":4,"created":"2020-10-30 04:43:57.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/84/760284/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/760284/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/760284/4"}}},"commit":{"parents":[{"commit":"d21d37d6285fa949f286c86ccbccd8fc869bb5bf","subject":"Merge \"Release note for mysql 8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d21d37d6285fa949f286c86ccbccd8fc869bb5bf"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 04:32:28.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-30 04:43:48.000000000","tz":660},"subject":"Support ram quota","message":"Support ram quota\n\nSet default to -1 so this is backwards compatible. Existing\ninstallations will need to manully backfill quote usage for this\nto work as expected.\n\nStory: 2008293\nTask: 41172\n\nChange-Id: I455477a2e7a00f0d132971a2a684352967ac19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/704b40cdcedca36b878dfd03a2fe605aa9cf33d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/704b40cdcedca36b878dfd03a2fe605aa9cf33d3"}]},"branch":"refs/heads/master"},"d69c99db9564e014f7a7987bff744bece35ca38b":{"kind":"NO_CHANGE","_number":5,"created":"2020-10-30 04:54:23.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/84/760284/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/760284/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/760284/5"}}},"commit":{"parents":[{"commit":"d21d37d6285fa949f286c86ccbccd8fc869bb5bf","subject":"Merge \"Release note for mysql 8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d21d37d6285fa949f286c86ccbccd8fc869bb5bf"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 04:32:28.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-30 04:54:13.000000000","tz":660},"subject":"Support ram quota","message":"Support ram quota\n\nSet default to -1 so this is backwards compatible. Existing\ninstallations will need to manully backfill quote usage for this\nto work as expected.\n\nStory: 2008293\nTask: 41172\n\nChange-Id: I455477a2e7a00f0d132971a2a684352967ac19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d69c99db9564e014f7a7987bff744bece35ca38b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d69c99db9564e014f7a7987bff744bece35ca38b"}]},"branch":"refs/heads/master"},"378178f21ba849e251add89300b119d092e787df":{"kind":"REWORK","_number":6,"created":"2020-10-30 04:55:28.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/84/760284/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/760284/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/760284/6"}}},"commit":{"parents":[{"commit":"69c8467854795fd3a01d3fd0865aceb2c1806219","subject":"Return instances by created order","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69c8467854795fd3a01d3fd0865aceb2c1806219"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 04:32:28.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-30 04:55:19.000000000","tz":660},"subject":"Support ram quota","message":"Support ram quota\n\nSet default to -1 so this is backwards compatible. Existing\ninstallations will need to manully backfill quote usage for this\nto work as expected.\n\nStory: 2008293\nTask: 41172\n\nChange-Id: I455477a2e7a00f0d132971a2a684352967ac19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/378178f21ba849e251add89300b119d092e787df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/378178f21ba849e251add89300b119d092e787df"}]},"branch":"refs/heads/master"},"332879041f671156722b64ddd67d7e4cdcc7b2e2":{"kind":"REWORK","_number":7,"created":"2020-11-03 20:58:41.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/84/760284/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/760284/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/760284/7"}}},"commit":{"parents":[{"commit":"71d99e2c56a0b6decfdf845ed40b35788645fffb","subject":"Add doc for running functional test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d99e2c56a0b6decfdf845ed40b35788645fffb"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 04:32:28.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-03 20:58:29.000000000","tz":660},"subject":"Support ram quota","message":"Support ram quota\n\nSet default to -1 so this is backwards compatible. Existing\ninstallations will need to manully backfill quote usage for this\nto work as expected.\n\nStory: 2008293\nTask: 41172\n\nChange-Id: I455477a2e7a00f0d132971a2a684352967ac19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/332879041f671156722b64ddd67d7e4cdcc7b2e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/332879041f671156722b64ddd67d7e4cdcc7b2e2"}]},"branch":"refs/heads/master"},"bb164d5df89a372f00a8d19472ed75fcf55ee68e":{"kind":"REWORK","_number":8,"created":"2020-11-10 23:05:26.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/84/760284/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/760284/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/760284/8"}}},"commit":{"parents":[{"commit":"78772cef68887a59adbde42057ef43d586e28b85","subject":"Merge \"Make guest agent api changes backwards compatible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78772cef68887a59adbde42057ef43d586e28b85"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 04:32:28.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-10 23:05:16.000000000","tz":660},"subject":"Support ram quota","message":"Support ram quota\n\nSet default to -1 so this is backwards compatible. Existing\ninstallations will need to manully backfill quote usage for this\nto work as expected.\n\nStory: 2008293\nTask: 41172\n\nChange-Id: I455477a2e7a00f0d132971a2a684352967ac19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bb164d5df89a372f00a8d19472ed75fcf55ee68e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bb164d5df89a372f00a8d19472ed75fcf55ee68e"}]},"branch":"refs/heads/master"},"5be23d1b20bf96f9605b4ae0cf6d7eb1d17d7a2b":{"kind":"REWORK","_number":9,"created":"2020-11-11 09:51:35.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/84/760284/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/84/760284/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/84/760284/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/84/760284/9"}}},"commit":{"parents":[{"commit":"78772cef68887a59adbde42057ef43d586e28b85","subject":"Merge \"Make guest agent api changes backwards compatible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78772cef68887a59adbde42057ef43d586e28b85"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-10-29 04:32:28.000000000","tz":660},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-11 09:51:26.000000000","tz":0},"subject":"Support ram quota","message":"Support ram quota\n\nSet default to -1 so this is backwards compatible. Existing\ninstallations will need to manully backfill quote usage for this\nto work as expected.\n\nStory: 2008293\nTask: 41172\n\nChange-Id: I455477a2e7a00f0d132971a2a684352967ac19b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5be23d1b20bf96f9605b4ae0cf6d7eb1d17d7a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5be23d1b20bf96f9605b4ae0cf6d7eb1d17d7a2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
