)]}'
{"id":"openstack%2Ftrove~760475","triplet_id":"openstack%2Ftrove~master~Ida407c05ad1026820be71ddfb84a97bcb3418df9","project":"openstack/trove","branch":"master","topic":"tox","hashtags":[],"change_id":"Ida407c05ad1026820be71ddfb84a97bcb3418df9","subject":"Update TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2020-10-30 01:06:31.000000000","updated":"2020-10-30 07:32:18.000000000","submitted":"2020-10-30 07:30:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760475-1604043036781-b6ee05d8","meta_rev_id":"cb5ad41de5e06863d915bdeb82a0189dac80b5f8","_number":760475,"virtual_id_number":760475,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-30 07:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"recommended":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"all":[{"value":1,"date":"2020-10-30 03:21:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":2,"date":"2020-10-30 05:45:00.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2020-10-30 05:45:00.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 03:21:05.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-10-30 05:45:00.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-10-30 07:30:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1231cba840c996275de8d6db2887d475219378a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-10-30 01:06:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d45d425c791ea7535c574eec4d9a2bd3ff267ae","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-10-30 03:21:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1e56acc9f009e6d14a786bc1a588ef0afe0dde6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 03:53:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2aca2373eaa34c2b86b7fee66d4916f0 : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/781ad2dc7ddc4001a7c504099c6a90fa : FAILURE in 8m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1faa9cd5b2e64a26b8eb340aedce3000 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841f55fe126c4e679280f86eeeec2ec2 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3960f1e883854d1894c33da4647c075d : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d4c0279e4344728be273770b4e7d0bf : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff312932a1bf4aec8b11a66500795fb4 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e84fbfbf64144b058dae34228eef0942 : SUCCESS in 5m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/19035646b31a4b5ba4fa858b6727c8ab : SUCCESS in 5m 49s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2d26d9a1cb624fc08b5d88fdde3febf6 : TIMED_OUT in 2h 13m 15s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4140d1ee26f243bd809b4d7edfafcd19 : RETRY_LIMIT in 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31965c96698c9333f769690be85746beb22c1903","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-10-30 05:45:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"b00fbc8c770ddbe104a27d2656be2549ea6801e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 05:45:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb7d08f10172fae30b4689c693632f012d7a9987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 07:30:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/64d6034ad36a4c779ccba1cfc31d7c04 : SUCCESS in 7m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c4543e42bca48b8ab8a33fc436e87f0 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e830adabf3e34c13bb3a22b8ef23993f : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21c66f8bc96b4f75a59dfffc06dd12de : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/922b0fb2f9f54e799de3ddb289822a2a : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaba83c75fca4501925ecc5653656666 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90f81074766c4c8f8c29a39767185043 : SUCCESS in 4m 49s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a1d16a75ce18454895f4790836141961 : FAILURE in 1h 43m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7bf351984a1538713e408bc10b219893802a0f7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 07:30:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"86e40f0219a10aa0f93c33ffd554f7d7773f5e66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 07:32:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/5b35992878fe4a2a9f6a21d6c69322dd : SUCCESS in 54s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ae04a527144f6f8c76f747e672c873 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0636f51ca54c4a3d9191a4cc18e48bc2 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3753c70225a40bd71276715fd9bc13ce2fb9ac6a","revisions":{"3753c70225a40bd71276715fd9bc13ce2fb9ac6a":{"kind":"REWORK","_number":1,"created":"2020-10-30 01:06:31.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/75/760475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/75/760475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/760475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/760475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/75/760475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/75/760475/1"}}},"commit":{"parents":[{"commit":"d21d37d6285fa949f286c86ccbccd8fc869bb5bf","subject":"Merge \"Release note for mysql 8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d21d37d6285fa949f286c86ccbccd8fc869bb5bf"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-10-30 01:04:53.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-10-30 01:04:53.000000000","tz":480},"subject":"Update TOX_CONSTRAINTS_FILE","message":"Update TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\n-https://zuul-ci.org/docs/zuul-jobs/python-roles.html#rolevar-tox.tox_constraints_file\nThis allows to use lower-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\n[1] https://review.opendev.org/#/c/722814/\n\nChange-Id: Ida407c05ad1026820be71ddfb84a97bcb3418df9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3753c70225a40bd71276715fd9bc13ce2fb9ac6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3753c70225a40bd71276715fd9bc13ce2fb9ac6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
