)]}'
{"id":"openstack%2Ftrove~761063","triplet_id":"openstack%2Ftrove~master~I77b72a965153e9583d93dee1f3a77d01a57e4ca2","project":"openstack/trove","branch":"master","topic":"flavor-back","hashtags":[],"change_id":"I77b72a965153e9583d93dee1f3a77d01a57e4ca2","subject":"Revert \"Remove flavor API\"","status":"MERGED","created":"2020-11-03 07:22:58.000000000","updated":"2020-11-05 06:05:17.000000000","submitted":"2020-11-05 06:03:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761063-1604556226884-4fbb7558","meta_rev_id":"52cebd434642cf1d60db50628f6ae97dbd8fbcd2","_number":761063,"virtual_id_number":761063,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-05 06:03:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"recommended":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"all":[{"value":1,"date":"2020-11-05 02:46:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":2,"date":"2020-11-05 04:10:04.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2020-11-05 04:10:04.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 04:10:04.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-11-05 06:03:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"284dc4b5bbc66137b94d1112bced10b199a3ddbb","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-03 07:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa6e48ac963e783af91ba901b4d26213392d1bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:51:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8849ce524e7147f696e2756abd4e6e38 : SUCCESS in 5m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd51837fa37d4312abedc5ae520c0cb0 : FAILURE in 8m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c8ed09440ee4ae6bbbe93e65794e3a4 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a9c53a890b49519589ccd55a36a09a : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcfd3dcf0b0549d8a6bed05b2e01cfe4 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e12b5d89e30949128947d2f72e87b849 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c4cac39a27e433cbe9ba787719272ea : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2f061b0d57484abb82f6d43c3e3663 : SUCCESS in 5m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2768f0971aeb4ceaaddeda2ef4388b33 : SUCCESS in 5m 40s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/98c4145b6f24459696af0d9965132cd9 : FAILURE in 1h 52m 01s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec103d511c30493a8579cc244c76822e : SUCCESS in 1h 08m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a5bbf4f04f25ddd3b6212ecb6e5a924adc2eb75","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-03 19:44:11.000000000","message":"Patch Set 1: Code-Review-1\n\nCan we simply only add the API back but not the test and API changes? After adding the API back, we will immediately mark it deprecated.\n\nI am still curious what are you relying on trove flavor API for instead of Nova API?","accounts_in_message":[],"_revision_number":1},{"id":"91edfe9fa4be354b8a78d773cdb6b043aacbdc02","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-03 20:46:06.000000000","message":"Patch Set 1:\n\n\u003e Can we simply only add the API back but not the test and API\n \u003e changes? After adding the API back, we will immediately mark it\n \u003e deprecated.\n\nMaybe this could be done in a separate review to mark it as deprecated?\n \u003e \n \u003e I am still curious what are you relying on trove flavor API for\n \u003e instead of Nova API?\n\nWe can\u0027t expect all our users to upgrade their clients, it\u0027s just not a good idea to break an API. We don\u0027t know how our users are using trove, they might have created their own client in go for all we know","accounts_in_message":[],"_revision_number":1},{"id":"852b5ccb13836c59de0e308fb9ed67a1fce6db80","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-03 21:04:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e0d99ccc195e6eeb9693faa179093be50dbaffde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 23:08:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f573072be13e47e8b6f4e88117b2369f : FAILURE in 7m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ae11fc12f9846599c4b56b096dc51df : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bb528870a1b4c40884a057c1cc56e08 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18fe359cd9f24151820a5f06ea1d29dd : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3304c696e684f96a2ca19faf0a6dd8a : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d2b47af3c0245239573a234d5aa32d5 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd06bf9e3974493b95fdc68882ee00fd : SUCCESS in 5m 41s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e6082f8622294fc78ca689688e0fa02d : SUCCESS in 5m 09s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c417d841fc7c4aca952bdeaa872f1254 : FAILURE in 1h 51m 25s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7245a9edbe24e9282cec5857f556154 : FAILURE in 1h 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af219c858b00873c640423497e7d65214a9b59e4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-04 04:06:56.000000000","message":"Patch Set 2: Code-Review-1\n\nfunctional test failed with http://paste.openstack.org/show/799675/, you could either change the test (e.g. remove attribute check) or remove \u0027links\u0027 from flavor attribute.","accounts_in_message":[],"_revision_number":2},{"id":"9ce5b0a2543144da55317b389033b365505c3464","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-04 06:26:11.000000000","message":"Patch Set 2:\n\n\u003e functional test failed with http://paste.openstack.org/show/799675/,\n \u003e you could either change the test (e.g. remove attribute check) or\n \u003e remove \u0027links\u0027 from flavor attribute.\n\nMaybe it\u0027s better to just revert it cleanly and include all the tests and the docs. I don\u0027t see how we can easily remove an API endpoint at all without breaking things for users.","accounts_in_message":[],"_revision_number":2},{"id":"142196ad5ad1e0594edcad8b6783ce5398784d20","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-05 00:10:36.000000000","message":"Patch Set 2:\n\n\u003e Maybe it\u0027s better to just revert it cleanly and include all the\n \u003e tests and the docs. I don\u0027t see how we can easily remove an API\n \u003e endpoint at all without breaking things for users.\n\nI will fix the test for you. BTW, have you run the functional test even if reverting all the stuff?","accounts_in_message":[],"_revision_number":2},{"id":"77ea2402c51295f06a92810f7e80755e6115aa17","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-05 01:31:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"94906e51c9d6cecf3ea7203c8d046529d3f2ecc0","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-05 02:46:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"58ca95d3d44631eff0d13f03339708bcdff74fe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 03:20:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47f078bcd3dc481694575d820781d839 : FAILURE in 6m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62e22aef985646d080b3b77ff3764d09 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0746b434db7545719f3aa9429d28c052 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dae5fdaabd834244b8833514f97a4bb8 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3499fb149d37467fa20892e6b6de3511 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab0d65ddf8104428b7a054988b4ffba5 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15eda5f4c8264dab95eef8271d049411 : SUCCESS in 4m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a872b5de59e54e81bd631ac6dae7539f : SUCCESS in 5m 03s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/acad9caa63b44262a6b81789c716dd33 : FAILURE in 1h 46m 50s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8590e08b6bc74bf49b42401a11abc388 : SUCCESS in 59m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e41fbabfd03231281c9874e09511c594279b1e86","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-05 04:10:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cdd272e5cae5e4627141460c8070ce05dc283ed9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-05 04:10:17.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit cab5015af551f70159129e18f890ad3e05359687","accounts_in_message":[],"_revision_number":3},{"id":"c4050d8b4227b3916e5a80573b20e7f28058df4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 04:10:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d9fc0f7f7ae87a4b440dc3b103d86653baf998e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 06:03:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23750615ab7a4a2c99b8e16c500fa4c2 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60c0b205a13a4f4eaad057683f214f17 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/841ee0b9c64c45d5a4f6f84cf347489c : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34517c7cb770401d9c824540c145212b : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444ed381b9694187aed8113c5bd49b90 : SUCCESS in 6m 23s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3e4e81191d964ef89292140c600d2c2e : FAILURE in 1h 50m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d3ea1eba9914bd5a248c65f0e4c7709a1d92594","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 06:03:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"85984778f95e4cb2929748ad30c362dc56e91f1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 06:05:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427c43974a124f7e9104663970e2b6e3 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3aa0c32815acba5c6b27f4e29a66523ba9366c90","revisions":{"811fc23429b72c4595df799a63ccf44b355826a3":{"kind":"REWORK","_number":1,"created":"2020-11-03 07:22:58.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/63/761063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/63/761063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/761063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/761063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/761063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/63/761063/1"}}},"commit":{"parents":[{"commit":"71d99e2c56a0b6decfdf845ed40b35788645fffb","subject":"Add doc for running functional test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d99e2c56a0b6decfdf845ed40b35788645fffb"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-03 07:16:43.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-03 07:22:55.000000000","tz":660},"subject":"Revert \"Remove flavor API\"","message":"Revert \"Remove flavor API\"\n\nStory: 2008309\nTask: 41201\n\nThis reverts commit 910519127d8fb9880863eb219df68a574d00df6a.\n\nChange-Id: I77b72a965153e9583d93dee1f3a77d01a57e4ca2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/811fc23429b72c4595df799a63ccf44b355826a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/811fc23429b72c4595df799a63ccf44b355826a3"}]},"branch":"refs/heads/master"},"e73ed7231e01b6c44991b66cd17644805a01c09b":{"kind":"REWORK","_number":2,"created":"2020-11-03 21:04:04.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/63/761063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/63/761063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/761063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/761063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/761063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/63/761063/2"}}},"commit":{"parents":[{"commit":"71d99e2c56a0b6decfdf845ed40b35788645fffb","subject":"Add doc for running functional test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d99e2c56a0b6decfdf845ed40b35788645fffb"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-03 07:16:43.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-03 21:03:13.000000000","tz":660},"subject":"Revert \"Remove flavor API\"","message":"Revert \"Remove flavor API\"\n\nStory: 2008309\nTask: 41201\n\nThis reverts commit 910519127d8fb9880863eb219df68a574d00df6a.\n\nChange-Id: I77b72a965153e9583d93dee1f3a77d01a57e4ca2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e73ed7231e01b6c44991b66cd17644805a01c09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e73ed7231e01b6c44991b66cd17644805a01c09b"}]},"branch":"refs/heads/master"},"3aa0c32815acba5c6b27f4e29a66523ba9366c90":{"kind":"REWORK","_number":3,"created":"2020-11-05 01:31:51.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/63/761063/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/63/761063/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/761063/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/761063/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/761063/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/63/761063/3"}}},"commit":{"parents":[{"commit":"71d99e2c56a0b6decfdf845ed40b35788645fffb","subject":"Add doc for running functional test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d99e2c56a0b6decfdf845ed40b35788645fffb"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-03 07:16:43.000000000","tz":660},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-05 01:31:27.000000000","tz":780},"subject":"Revert \"Remove flavor API\"","message":"Revert \"Remove flavor API\"\n\nStory: 2008309\nTask: 41201\n\nThis reverts commit 910519127d8fb9880863eb219df68a574d00df6a.\n\nChange-Id: I77b72a965153e9583d93dee1f3a77d01a57e4ca2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3aa0c32815acba5c6b27f4e29a66523ba9366c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3aa0c32815acba5c6b27f4e29a66523ba9366c90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
