)]}'
{"id":"openstack%2Ftrove~761250","triplet_id":"openstack%2Ftrove~master~I62ea46991c7398fba4e5027d5170038624d5aec9","project":"openstack/trove","branch":"master","topic":"volume","hashtags":[],"change_id":"I62ea46991c7398fba4e5027d5170038624d5aec9","subject":"Pass availability zone through to volume creation.","status":"MERGED","created":"2020-11-03 23:45:27.000000000","updated":"2020-11-10 01:14:15.000000000","submitted":"2020-11-10 01:12:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761250-1604970767594-4b8f775d","meta_rev_id":"0e020311682d0f4053e7e21f4219a920b1ab4e4e","_number":761250,"virtual_id_number":761250,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-10 01:12:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-11-09 23:11:52.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-09 23:11:52.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-09 23:11:52.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-11-10 01:12:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"045eaab98ae4e09c3b832add697830f16f992ee1","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-03 23:45:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f662b81fd3ddb0eaa558c87070a3d1ec2f8b81c4","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-04 00:18:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"917b1992a8737c1f247745002cb4c493dc195331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 01:56:37.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/197985fe932b4f548494ff5520c3e8a1 : FAILURE in 11m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6d59786a45244f7a331bffb6825ff73 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1efda9fb1964601945006f79604e93b : FAILURE in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6df077f0ae0445d7be1a561018823bc0 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cae512bdafd64400823a8d248abf6d62 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90730e7813084ac3b01f40f4f7114f45 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e72104f683747a5be5060601a27f9ec : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55b318b308d64cea80d8147844215d19 : SUCCESS in 5m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/2f0022eb21c743f4a818c7f2487bd8f1 : SUCCESS in 5m 28s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a97dd529359d438d9df143393c54703b : FAILURE in 1h 36m 47s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31be38a0b35b44dfbbfacd94f1ee4e03 : FAILURE in 1h 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa9684f9d26cd279a4bf329d19c4fc5659517237","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-05 01:07:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"03044c0f7abfa1b7fe055311402dd93e95c98c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 03:26:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6da761bced424a578386552bef24aa77 : FAILURE in 5m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82c6755757e948aaadc5a4ddd2156ad8 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ccc0675d26f41c69a991d13a80379b6 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2c9a42f9c65488381e4cab8f5514659 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40e19aa2cf764f14a224a32ca06df0ad : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/181f737f3fdd4db1a9ce0e78c4c7e690 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d88fba8ece423a837d3e7e2a56da83 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cada5a5c31f0499285f0093d6bd5d2c0 : SUCCESS in 5m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6d3e118d16114cc3b5bb772542cd5c50 : SUCCESS in 5m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5470216b64df425081c8bcc481c309f4 : FAILURE in 2h 12m 48s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a535bb877d1a48dc98bac042af6ec002 : SUCCESS in 1h 02m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8426a15f8da6b9984d513840c360d8b3751e7db9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-05 20:41:54.000000000","message":"Patch Set 3:\n\nHow about creating volume using the az param by default without introducing an extra config option?","accounts_in_message":[],"_revision_number":3},{"id":"b50c9ee618763739d41dcddcc6612f41d8c64202","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-06 04:27:11.000000000","message":"Patch Set 3:\n\n\u003e How about creating volume using the az param by default without\n \u003e introducing an extra config option?\n\nDoing it by default would be a backwards incompatible change. Some sites don\u0027t have cinder AZs and they use the 1 AZ for multiple nova AZs","accounts_in_message":[],"_revision_number":3},{"id":"a5aac563bb3f85843c017a2f7990c7d4382abeda","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-08 21:56:38.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nI will run functional test on my local.","accounts_in_message":[],"_revision_number":3},{"id":"db8640c4f42acde9f07dd8f88fcaa8c398073f97","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-08 22:56:43.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a6667070f8fa3956fe1c70fbdba759a93e2dec29","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-08 23:01:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"598b41487bfabecb7500d41618b7fa6613627651","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-09 00:52:39.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"571a577e7db140e9b39691676162629b5eb04ab0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-09 00:54:02.000000000","message":"Patch Set 5:\n\nHi sorrison, do you want this patch to be backported to Victoria?","accounts_in_message":[],"_revision_number":5},{"id":"3a6a4efb23271b0f0d9634dc24ab96124f95546d","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-11-09 00:58:44.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 7cfa7d9951b7cf4f39122d7759eef37ece4db74c","accounts_in_message":[],"_revision_number":5},{"id":"b571ab91d1a799f48b92b10fe465f9be2d59a4cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 02:43:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6ba2750fa834a2594b41e0579f63520 : FAILURE in 7m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/587bdda5440e40ee9030d6702cb116c9 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fee077dc4393431b8702e461d38b32d2 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1bfeadf09c047d5b07efc78863c109e : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/342afdd54564439f8d9c794577670c58 : SUCCESS in 11m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b0f041b23dd4d74a82dfbdb325aee6b : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5d515077217425185aa80000e20f17d : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37715beed6e640998b75236cace1998f : SUCCESS in 4m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/013ae1f3c0eb482d937e00bb68feec46 : SUCCESS in 5m 42s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3d3489aeb4ee4d9595d0618b31cbc824 : FAILURE in 1h 49m 11s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9c6bc7311f34d2a9193b7d19b5520b3 : FAILURE in 1h 18m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f1544cf050f9c49caa1c0ea45128ca54b658712e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-09 23:11:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b75778af43f356d8cf9a1000fcaaad14e3e28c87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 23:12:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6b6a0bf03865578ecfa83c4faf657a1e3fe8264a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 01:12:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"19214f6492361dfbe511ee505f5b032062f0ca14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 01:12:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d033e6f52cf4747aecd58a2e73f1712 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa6b7c92c934a96a37acaa7cd1627be : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/424e974f3df14a6b852ae1cf4cc0bf96 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e4c3d424de64a4795b26205ca24495c : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e08490af455e4c47a21258682bbfd5a5 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afdf97f133f34d31b84aea277fb4cc5f : SUCCESS in 5m 16s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2a7f932fdb1d4a729b309909a9adacec : FAILURE in 1h 49m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ed10753c12125b5d349b4f277d1071b981732798","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 01:14:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1500e310abc4bbfad1303ee33a0ae63 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/762e4119ab8c40e6ad2b4f796b63afd7 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e8287dfda9159acffc434432a31b6c0b671aa83e","revisions":{"31e7733033f8b87e62fb298a0b0e5463a77f5b9a":{"kind":"REWORK","_number":1,"created":"2020-11-03 23:45:27.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/50/761250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/50/761250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/50/761250/1"}}},"commit":{"parents":[{"commit":"71d99e2c56a0b6decfdf845ed40b35788645fffb","subject":"Add doc for running functional test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d99e2c56a0b6decfdf845ed40b35788645fffb"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2016-09-19 01:41:22.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-03 23:44:10.000000000","tz":660},"subject":"Pass availability zone through to volume creation.","message":"Pass availability zone through to volume creation.\n\nNew config option `use_az_for_volume` (defaults to false)\nensures that the instance and volume will be created in the same\navailability zone.\n\nStory: 2008313\nTask: 41208\n\nChange-Id: I62ea46991c7398fba4e5027d5170038624d5aec9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/31e7733033f8b87e62fb298a0b0e5463a77f5b9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/31e7733033f8b87e62fb298a0b0e5463a77f5b9a"}]},"branch":"refs/heads/master"},"76174d71cd41cda4929be7e6e8ecf2eef561535a":{"kind":"REWORK","_number":2,"created":"2020-11-04 00:18:03.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/50/761250/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/50/761250/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/50/761250/2"}}},"commit":{"parents":[{"commit":"71d99e2c56a0b6decfdf845ed40b35788645fffb","subject":"Add doc for running functional test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d99e2c56a0b6decfdf845ed40b35788645fffb"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2016-09-19 01:41:22.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-04 00:17:41.000000000","tz":660},"subject":"Pass availability zone through to volume creation.","message":"Pass availability zone through to volume creation.\n\nNew config option `use_az_for_volume` (defaults to false)\nensures that the instance and volume will be created in the same\navailability zone.\n\nStory: 2008313\nTask: 41208\n\nChange-Id: I62ea46991c7398fba4e5027d5170038624d5aec9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76174d71cd41cda4929be7e6e8ecf2eef561535a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/76174d71cd41cda4929be7e6e8ecf2eef561535a"}]},"branch":"refs/heads/master"},"1a5615de05df2b43c480d3fafc5f69fb0db37dec":{"kind":"REWORK","_number":3,"created":"2020-11-05 01:07:26.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/50/761250/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/50/761250/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/50/761250/3"}}},"commit":{"parents":[{"commit":"71d99e2c56a0b6decfdf845ed40b35788645fffb","subject":"Add doc for running functional test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d99e2c56a0b6decfdf845ed40b35788645fffb"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2016-09-19 01:41:22.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-05 01:07:15.000000000","tz":660},"subject":"Pass availability zone through to volume creation.","message":"Pass availability zone through to volume creation.\n\nNew config option `use_az_for_volume` (defaults to false)\nensures that the instance and volume will be created in the same\navailability zone.\n\nStory: 2008313\nTask: 41208\n\nChange-Id: I62ea46991c7398fba4e5027d5170038624d5aec9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a5615de05df2b43c480d3fafc5f69fb0db37dec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a5615de05df2b43c480d3fafc5f69fb0db37dec"}]},"branch":"refs/heads/master"},"75dad0a42a5dfc59d8820f67991681a6e24a252c":{"kind":"REWORK","_number":4,"created":"2020-11-08 23:01:18.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/50/761250/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/50/761250/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/50/761250/4"}}},"commit":{"parents":[{"commit":"59a972b312c8d25ffb46b4a994374490731747ea","subject":"Fix upgrading instance datastore version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59a972b312c8d25ffb46b4a994374490731747ea"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2016-09-19 01:41:22.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-11-08 23:01:07.000000000","tz":660},"subject":"Pass availability zone through to volume creation.","message":"Pass availability zone through to volume creation.\n\nNew config option `use_az_for_volume` (defaults to false)\nensures that the instance and volume will be created in the same\navailability zone.\n\nStory: 2008313\nTask: 41208\n\nChange-Id: I62ea46991c7398fba4e5027d5170038624d5aec9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/75dad0a42a5dfc59d8820f67991681a6e24a252c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/75dad0a42a5dfc59d8820f67991681a6e24a252c"}]},"branch":"refs/heads/master"},"e8287dfda9159acffc434432a31b6c0b671aa83e":{"kind":"REWORK","_number":5,"created":"2020-11-09 00:52:39.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/50/761250/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/50/761250/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/761250/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/50/761250/5"}}},"commit":{"parents":[{"commit":"59a972b312c8d25ffb46b4a994374490731747ea","subject":"Fix upgrading instance datastore version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/59a972b312c8d25ffb46b4a994374490731747ea"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2016-09-19 01:41:22.000000000","tz":600},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-09 00:50:49.000000000","tz":0},"subject":"Pass availability zone through to volume creation.","message":"Pass availability zone through to volume creation.\n\nNew config option `enable_volume_az` (defaults to false)\nensures that the instance and volume will be created in the same\navailability zone.\n\nStory: 2008313\nTask: 41208\n\nChange-Id: I62ea46991c7398fba4e5027d5170038624d5aec9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8287dfda9159acffc434432a31b6c0b671aa83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8287dfda9159acffc434432a31b6c0b671aa83e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
