)]}'
{"id":"openstack%2Ftrove~762731","triplet_id":"openstack%2Ftrove~master~Idacd5b0728edd792fd3aa8206c84bc837e9a88d0","project":"openstack/trove","branch":"master","topic":"delete-instance","hashtags":[],"change_id":"Idacd5b0728edd792fd3aa8206c84bc837e9a88d0","subject":"Do not check server status when waiting for instance removed","status":"MERGED","created":"2020-11-14 10:10:56.000000000","updated":"2020-11-14 23:31:09.000000000","submitted":"2020-11-14 23:30:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"762731-1605396601094-bced994f","meta_rev_id":"3c3db95ba2267ce829651a294aba07448c940369","_number":762731,"virtual_id_number":762731,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-14 23:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-11-14 21:33:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-11-14 21:33:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-14 23:30:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e0fb17cf711f1d486aa1a7c3cf6a30a4b82df6e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-14 10:10:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79ee201becc353e0af6facd40eed0342efd8f78c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-14 10:11:57.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/victoria as commit 698300d8ae1dfdf75f742c94c625a78a37ea069a","accounts_in_message":[],"_revision_number":1},{"id":"762397f1fb7bd62ef9b6972ec1f1bef3a20b93e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 12:31:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a1deb81a8214befbb4d54d611fd3eb1 : FAILURE in 6m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca30a31c43fb465daf87c61f25ee61c1 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc55432577414ea7bbfa7f80e931b2d9 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bfe795ea0614392bc3fb2e36e2f29f4 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/025aed63cac14f88bd407826175b4994 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6e48579f5764cb98a78a038308283db : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/745187e8663d4896965bb328db0fd530 : SUCCESS in 5m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c281b183cee84c6c97a546f8ecfb1146 : SUCCESS in 5m 51s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/42071cbcff0247248faaf3db66475163 : TIMED_OUT in 2h 12m 41s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d8859c1e7834b1bb043edeb9c650ffc : FAILURE in 1h 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73c646b45d93b5f09d31a6f4988be93540d4c6e9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-14 21:33:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5026eaf83f872651e542d5f2a3600220624b1ec8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 21:33:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"773931b15db040bcfe96ea800eadeed6b40437cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 23:30:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4c99c84e3a1407a956336ba2abac264 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44bc472bc3c14e838ffbe1b242813a00 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a37fab9f1034736a34d834e4265cf84 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/977bd18a9ad5476694107be0960057f4 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781b9593b36249c58c76f43dab47960c : SUCCESS in 4m 55s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e748f45a63f9475494e51b348ab94888 : FAILURE in 1h 50m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87de50ee5f3c8a856edf28420e4710978edc143a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 23:30:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"08a58891fbb743c87429bed8e992750d28a30126","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-14 23:31:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560187bfcd15479288670f00b7ffd52f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a05b6ee13f59054a36e467453f953594dcdde643","revisions":{"a05b6ee13f59054a36e467453f953594dcdde643":{"kind":"REWORK","_number":1,"created":"2020-11-14 10:10:56.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/31/762731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/31/762731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/762731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/762731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/31/762731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/31/762731/1"}}},"commit":{"parents":[{"commit":"7a40914de29814e9db6254c36b889a8c890e565c","subject":"Merge \"Support ram quota\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7a40914de29814e9db6254c36b889a8c890e565c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-14 10:10:41.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-14 10:10:41.000000000","tz":780},"subject":"Do not check server status when waiting for instance removed","message":"Do not check server status when waiting for instance removed\n\nChange-Id: Idacd5b0728edd792fd3aa8206c84bc837e9a88d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a05b6ee13f59054a36e467453f953594dcdde643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a05b6ee13f59054a36e467453f953594dcdde643"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
