)]}'
{"id":"openstack%2Ftrove~763259","triplet_id":"openstack%2Ftrove~master~Ieb0def81f2347c06a05eb354ad36b4fa807f7621","project":"openstack/trove","branch":"master","topic":"ds-version-4-configuration-params","hashtags":[],"change_id":"Ieb0def81f2347c06a05eb354ad36b4fa807f7621","subject":"Support datastore version number for creating configuration","status":"MERGED","created":"2020-11-18 20:36:22.000000000","updated":"2020-11-19 02:34:14.000000000","submitted":"2020-11-19 02:31:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763259-1605753071527-6db76d6a","meta_rev_id":"f26293e2ed66e68b957db2cdab40d79c2dfc61af","_number":763259,"virtual_id_number":763259,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-19 02:31:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-11-19 00:01:40.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-11-19 00:01:40.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-19 02:31:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d512eb0c667eb3aed6b7b982d2f81a7fb9cc0d8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-18 20:36:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65ca2822773b4edf20d76f0631c138d61f8a43a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 22:56:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21d6228147af49b18faff6ff9fedd9b8 : FAILURE in 7m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b6a7bad9e3747b087b0976660c7acdd : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b54b8947d0411a914f2f2221a14584 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/470b0476ce6b46f283bd1f54c9abd7e8 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb9499ef72f641c193a2f0c73eb76a26 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c2120a3fc9f4648b9337f3940ca87f0 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61708ed433674d7d9d0568351db8f662 : SUCCESS in 8m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d26f871734b48ee91bb29b58264290e : SUCCESS in 7m 53s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/89fc5bb78b974c1ebd9e969ec32423c1 : FAILURE in 2h 10m 39s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a231f3751e7b48538ed3052299782e26 : SUCCESS in 1h 15m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e6d3528fe8f54c7aa81e13ce50f6c47ac036487","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-19 00:01:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d5c3756a8ecc107346301a5f2d3f3e6a21326ac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 00:01:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af72a856b402659a6d6009dc947b26c2ae014545","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 02:31:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f099a18cb9a14938b45bb3320d76dda2 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d47f323b9d81475faf500ebfd5dc444f : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbc24533c149480ca7e45212d8a2d719 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26ed78a5387444f7a157bc307e8ef987 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794e4a319d51461cac261e386a11f2b9 : SUCCESS in 7m 01s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e7991a8fc4ef4fdf859bf6597b4db0c2 : TIMED_OUT in 2h 13m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf7b03540d8e60c719f86a1a5d9d5a2697b9c659","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 02:31:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"19289d9c05c8ef9ea0541839e383d332deec0a6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-19 02:34:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cecaf24d378b4d94ba7eaa7be831d33f : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d51b163e991f16acf2d542d99cbb2f2358feb9bc","revisions":{"d51b163e991f16acf2d542d99cbb2f2358feb9bc":{"kind":"REWORK","_number":1,"created":"2020-11-18 20:36:22.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/59/763259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/59/763259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/763259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/763259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/59/763259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/59/763259/1"}}},"commit":{"parents":[{"commit":"bf1ee5e8f29f286d2dee853db19912ea01ea4c78","subject":"Support datastore version number for creating instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bf1ee5e8f29f286d2dee853db19912ea01ea4c78"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-18 10:27:41.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-18 20:28:56.000000000","tz":780},"subject":"Support datastore version number for creating configuration","message":"Support datastore version number for creating configuration\n\n* Support \"version_number\" for creating configuration.\n* Fixed a bug for deleting datastore version, the config parameters should be\n  deleted first.\n* Support \"--version\" for trove-manage datastore_config_parameters\n  sub-commands.\n\nChange-Id: Ieb0def81f2347c06a05eb354ad36b4fa807f7621\nStory: 2008358\nTask: 41264\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d51b163e991f16acf2d542d99cbb2f2358feb9bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d51b163e991f16acf2d542d99cbb2f2358feb9bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
