)]}'
{"id":"openstack%2Ftrove~764303","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~I18607fb80beddf3f4008c7180626bd452edf52f5","project":"openstack/trove","branch":"stable/victoria","hashtags":[],"change_id":"I18607fb80beddf3f4008c7180626bd452edf52f5","subject":"No validate when perform eject replica source","status":"MERGED","created":"2020-11-26 23:04:26.000000000","updated":"2020-11-27 12:47:24.000000000","submitted":"2020-11-27 12:46:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764303","cherry_pick_of_change":762496,"cherry_pick_of_patch_set":4,"meta_rev_id":"3f4e4c0377b89042b4d20cb84ac3d1def6e6f354","_number":764303,"virtual_id_number":764303,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 12:46:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-11-27 10:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-11-27 10:36:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 23:04:26.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"state":"REVIEWER"},{"updated":"2020-11-27 01:07:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ccc1ea3b46c644d33aa7a07b4254fdd1e49498a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-26 23:04:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8972cf9806e2d900048af2d9f8bbb52320d1213c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-26 23:05:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7d480010cede4a2c1528cce853d3870369efd985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 01:07:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dc96d79c0c6471ba13582992025a539 : FAILURE in 9m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/deb0e1694626481e9369802f14d1363b : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2206f8e467640798114c5df1a4d08de : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b51cdecdb2644be78be09aaab40bae47 : SUCCESS in 9m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ad69aaba22747988aae70cc9b677e5e : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848677a31e5048a1aa2764d42dfbeb5e : SUCCESS in 5m 34s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ff4b4eb63b4946b2ac8e6620d2540c38 : SUCCESS in 6m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/092e90103ab546c8a64e4895c600f7fc : FAILURE in 2h 00m 13s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11ef6788b5e3405a96697f520f3e3295 : SUCCESS in 47m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9de16cbdcd93b2b3e35f870e68372d2f8461285b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-27 10:36:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1009351e7c14d98beebc294a20546efe12e4dfa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 10:37:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e76080aeb336feb7271e8bec43ae91345fb1ffbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 12:46:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/560dbf6ca2cd4d109a715207ccb9abc7 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e87f102420e4cdd813a17489cf121fe : SUCCESS in 9m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e13d66d3fb624c1eae7d6936d200934b : SUCCESS in 9m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42526e143e6e4b9eb58974fcf6384a2d : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae98879e28c74aa8a8c23a718d87da22 : SUCCESS in 5m 23s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2249647e44f2484aafb091e9344a869c : FAILURE in 1h 01m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"31c7907514ab1863b7a1a4086b06d0051737fd7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 12:46:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3f4e4c0377b89042b4d20cb84ac3d1def6e6f354","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 12:47:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0508de1079334152b63d08c309872544 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f4e377927a3b7e080804553be13aacb61c047c0","revisions":{"5d13ade15dcc578c464388dcafcf3b62bcb477a1":{"kind":"REWORK","_number":1,"created":"2020-11-26 23:04:26.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/03/764303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/764303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/764303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/764303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/764303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/764303/1"}}},"commit":{"parents":[{"commit":"64f3d52ffc05cd25414d6428e0e511a705fd06f6","subject":"Merge \"[docs]Fix wrong links in README\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/64f3d52ffc05cd25414d6428e0e511a705fd06f6"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-11-18 01:21:40.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-26 23:04:26.000000000","tz":0},"subject":"No validate when perform eject replica source","message":"No validate when perform eject replica source\n\nChange-Id: Id15496db1f1bb8bcae850608e01314ff7aafd40\n(cherry picked from commit 652b1e148301416c4e0879be1faa4be998d1ba41)\n\nChange-Id: I18607fb80beddf3f4008c7180626bd452edf52f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5d13ade15dcc578c464388dcafcf3b62bcb477a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5d13ade15dcc578c464388dcafcf3b62bcb477a1"}]},"branch":"refs/heads/stable/victoria"},"3f4e377927a3b7e080804553be13aacb61c047c0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-26 23:05:48.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/03/764303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/03/764303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/764303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/764303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/03/764303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/03/764303/2"}}},"commit":{"parents":[{"commit":"64f3d52ffc05cd25414d6428e0e511a705fd06f6","subject":"Merge \"[docs]Fix wrong links in README\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/64f3d52ffc05cd25414d6428e0e511a705fd06f6"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-11-18 01:21:40.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-26 23:05:44.000000000","tz":0},"subject":"No validate when perform eject replica source","message":"No validate when perform eject replica source\n\n(cherry picked from commit 652b1e148301416c4e0879be1faa4be998d1ba41)\n\nChange-Id: I18607fb80beddf3f4008c7180626bd452edf52f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f4e377927a3b7e080804553be13aacb61c047c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3f4e377927a3b7e080804553be13aacb61c047c0"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
