)]}'
{"id":"openstack%2Ftrove~764507","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~Ic4e2c79f48bfcaea806babbf50d4757a9c153132","project":"openstack/trove","branch":"stable/victoria","hashtags":[],"change_id":"Ic4e2c79f48bfcaea806babbf50d4757a9c153132","subject":"Get slave_pos to choose latest replica","status":"MERGED","created":"2020-11-30 09:02:25.000000000","updated":"2020-11-30 22:41:30.000000000","submitted":"2020-11-30 22:40:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764507","cherry_pick_of_change":763498,"cherry_pick_of_patch_set":1,"meta_rev_id":"38f25215e8a40d96c7a0c99e79a26b57f6524e43","_number":764507,"virtual_id_number":764507,"owner":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-30 22:40:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-11-30 20:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-11-30 20:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 09:31:40.000000000","updated_by":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-11-30 12:16:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"27bac9e3d36105d280e373dacf6444028a86a6da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"date":"2020-11-30 09:02:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"31b2c9785448676b8314c0de97115824ee48cfd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-30 09:56:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e6713c1c9037b3faaad75985d24b1f963edab698","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 12:16:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c17d820d8424c35a025e0b0a2a08598 : FAILURE in 6m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94ffb9c267274d2189591873a7c8b801 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9150df35d65d4113a3072f4735d684d4 : SUCCESS in 10m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47bba011403b43d1865e7981d560676a : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de04f7dea7fa4dc1936d13ecceedd195 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9413dcc0410b498f87692bcc032ed455 : SUCCESS in 7m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ab32a2e9d4ae473a84315fe5ef1f5407 : SUCCESS in 5m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6080f8facddf4317848bb90f74bc4977 : TIMED_OUT in 2h 12m 31s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41182792f3454f8bb066e566c53cff62 : SUCCESS in 1h 04m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3fcb7cf74e1a25b4cb1fe99009112a62d79de3db","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-11-30 20:31:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b1527a7cb341c8bcdbf48f7a58698add555e00ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 20:32:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4f505dda530660013028550d7d668bf1b447865e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 22:40:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6d5cefbcc1f42f498522f048cafadb0 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bea32458825b4c24b282a239b6a24d9b : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72f6f0444271405eb273d2fabc1ce8fb : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/207afc0213b34922bdd2aaf8ce8b8499 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61ee45b9646b455fa7497dd45cac475b : SUCCESS in 5m 52s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d857f21f1f464886b009ed4ef93767ce : FAILURE in 2h 06m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cfd5454419037d04113ea3529a1b237b10a0b1bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 22:40:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"38f25215e8a40d96c7a0c99e79a26b57f6524e43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 22:41:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e268a39904c4638b1aa245e958f2b2a : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"753252ab1c2248831403b04ebb4652d083c6278b","revisions":{"887235171ce5cf9c97d8357a0541201d5353c93c":{"kind":"REWORK","_number":1,"created":"2020-11-30 09:02:25.000000000","uploader":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"ref":"refs/changes/07/764507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/07/764507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/764507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/764507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/764507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/07/764507/1"}}},"commit":{"parents":[{"commit":"d6b35435e0114c663ab481680371de45acd99c2e","subject":"Use current slave_pos of slave to continue replicate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6b35435e0114c663ab481680371de45acd99c2e"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-11-20 03:00:09.000000000","tz":420},"committer":{"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","date":"2020-11-30 09:02:25.000000000","tz":0},"subject":"Get slave_pos to choose latest replica","message":"Get slave_pos to choose latest replica\n\nLatest replica have slave_pos biggest, so we just need use slave_pos to\ncompare replicas.\n\nChange-Id: Ic4e2c79f48bfcaea806babbf50d4757a9c153132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/887235171ce5cf9c97d8357a0541201d5353c93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/887235171ce5cf9c97d8357a0541201d5353c93c"}]},"branch":"refs/heads/stable/victoria"},"753252ab1c2248831403b04ebb4652d083c6278b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-30 09:56:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/07/764507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/07/764507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/764507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/764507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/764507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/07/764507/2"}}},"commit":{"parents":[{"commit":"d6b35435e0114c663ab481680371de45acd99c2e","subject":"Use current slave_pos of slave to continue replicate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d6b35435e0114c663ab481680371de45acd99c2e"}]}],"author":{"name":"Nguyen Thanh Cong","email":"ntcong1705@gmail.com","date":"2020-11-20 03:00:09.000000000","tz":420},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-11-30 09:56:58.000000000","tz":0},"subject":"Get slave_pos to choose latest replica","message":"Get slave_pos to choose latest replica\n\nLatest replica have slave_pos biggest, so we just need use slave_pos to\ncompare replicas.\n\nChange-Id: Ic4e2c79f48bfcaea806babbf50d4757a9c153132\n(cherry picked from commit df82af30f21b89097597bf471c2aa5a08d19386c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/753252ab1c2248831403b04ebb4652d083c6278b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/753252ab1c2248831403b04ebb4652d083c6278b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
