)]}'
{"id":"openstack%2Ftrove~765029","triplet_id":"openstack%2Ftrove~master~I3d8cfd57278ba426067f0631d63c02f2b73dcbf4","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I3d8cfd57278ba426067f0631d63c02f2b73dcbf4","subject":"Fixup zuul nodetype for nested","status":"MERGED","created":"2020-12-02 01:12:00.000000000","updated":"2020-12-03 01:08:01.000000000","submitted":"2020-12-03 01:03:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765029","meta_rev_id":"8a4ce4c4772f0fc2a7665778ab3da0b623088eed","_number":765029,"virtual_id_number":765029,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 01:03:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-12-02 22:23:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-12-02 22:23:30.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 04:13:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 22:23:30.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"30dfe7e7cc8782bc0f8da7024de096cdf9cd521a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-12-02 01:12:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8341b4fc1add7ece15162633ead107aa43540d7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 04:13:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf3902db62e84e04a1cc3b3509ab67e3 : FAILURE in 24m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68a8819c97b64f7f90d8ca606c5654a7 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ba7e43102b4a25a9a6ac8f17f7a6bf : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aac4eeadfd3d4cafa1decd1385bcc34c : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73f0238efed544b18e53ee9f13b80d34 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3f0a4a6fac7434fbfdfd3497ee94d31 : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc4701e7bab498ba864a3bc1ab9f25d : SUCCESS in 22m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f6da3062ff5148c68b0ce9abecee60b4 : SUCCESS in 24m 53s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/082de0d1245f4a68a63da2b1537ad842 : FAILURE in 2h 38m 31s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d683dcf687fb477c8050a9f7279f7c81 : SUCCESS in 1h 30m 51s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d6c9329576b64647b30ba03a09ae5287 : SUCCESS in 2h 53m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c450dd170d98ec6b083daf727ccd02ebdd94341","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-12-02 22:23:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nWill take a look at the tempest failure later on.","accounts_in_message":[],"_revision_number":1},{"id":"788ba05e736d82a87f3dc5c9d05656c12ae38bd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 22:24:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"467856343183b68bb2b81a62d967655a311ec596","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 01:03:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fb5ad9cf2fa4b49945ecb9281d7213c : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40057995bee4c3bab992513460eea97 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6af98e25772a4ad4b5799a9172f1db34 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dde78c03f9f475f9ce995643667317b : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0dde6c78b247309bb84d8613258aed : SUCCESS in 5m 59s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0ac168c1d3eb4aff97e8b6fe5a82fda9 : FAILURE in 2h 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51832a34825c9d551009254858171225b2094ecb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 01:03:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a4ce4c4772f0fc2a7665778ab3da0b623088eed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 01:08:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f27cb831f645d0945d9bfed26a051a : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25abf125ef1d751a34793ea7bcc92c03f926c152","revisions":{"25abf125ef1d751a34793ea7bcc92c03f926c152":{"kind":"REWORK","_number":1,"created":"2020-12-02 01:12:00.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/29/765029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/29/765029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/765029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/765029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/765029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/29/765029/1"}}},"commit":{"parents":[{"commit":"77c043062b2ea651684be2059fd7b27a29e6e684","subject":"Merge \"Get slave_pos to choose latest replica\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/77c043062b2ea651684be2059fd7b27a29e6e684"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-12-02 01:11:34.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-12-02 01:11:36.000000000","tz":660},"subject":"Fixup zuul nodetype for nested","message":"Fixup zuul nodetype for nested\n\nNeed tempest group to get tempest jobs to work\n\nChange-Id: I3d8cfd57278ba426067f0631d63c02f2b73dcbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/25abf125ef1d751a34793ea7bcc92c03f926c152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/25abf125ef1d751a34793ea7bcc92c03f926c152"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
