)]}'
{"id":"openstack%2Ftrove~766200","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~I8cd4554a20f89df92dbad4961d5a07a21f26fae8","project":"openstack/trove","branch":"stable/victoria","hashtags":[],"change_id":"I8cd4554a20f89df92dbad4961d5a07a21f26fae8","subject":"Disable openstack-lower-constraints-jobs","status":"MERGED","created":"2020-12-09 11:01:03.000000000","updated":"2020-12-09 22:46:26.000000000","submitted":"2020-12-09 22:21:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766200","meta_rev_id":"56dc28317dac66b5306425dfdbf308e3f7aaf365","_number":766200,"virtual_id_number":766200,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-09 22:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-12-09 19:36:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-12-09 19:36:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-09 13:19:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2a10ae46de51b8143178156a591a1e9c0bb6ddc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-12-09 11:01:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7012c39f9af8b377bf3269f859fa1337ba5e1c5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 13:19:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/193586ef91ca486db415ea60b0f2ac60 : FAILURE in 6m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ac9ff67e0c8438698a29befbc2fe59f : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7aae77859f142aeb46d1cb8775e721a : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05f5d401ad4e4eb094d0921b70983da8 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ded8b63074c4c9eb915c78050ef3c2e : SUCCESS in 5m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c5d0ccc55a1f49349bb5b41ccac3ec2e : SUCCESS in 7m 09s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3ec1ac2b52f9473c824f835d86e56c29 : FAILURE in 2h 11m 33s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30e33e391ddf4cfb91907bc13e077af0 : SUCCESS in 1h 54m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9a28b8590e804314d843b382595bab9a136c984","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-12-09 19:36:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"496797ccf57b527157eebf9b9b2505068f7ceef4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 19:37:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7fcbcd6bfd3ce7c85890acd02956d3d953b4e231","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 22:21:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a67cdf3cbf8c40e3ad608a69d2a94b37 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/104a9b9f50a04140bf4a74772105a16d : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eac7dfb3f9a04dddb206aef0f905fd73 : SUCCESS in 16m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92f4779fcc44964a7bd1bbb261a5b57 : SUCCESS in 16m 53s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4ef59721904b4936b86e50f3aa2e9b0f : TIMED_OUT in 2h 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a29036a5e306d40aab6c269ad883db0a793163f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 22:21:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"56dc28317dac66b5306425dfdbf308e3f7aaf365","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 22:46:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c968bb7adf7443abd01c530f24c548f : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3639befa7ef1c1e66d3c0a2960846dbe405f3b99","revisions":{"3639befa7ef1c1e66d3c0a2960846dbe405f3b99":{"kind":"REWORK","_number":1,"created":"2020-12-09 11:01:03.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/00/766200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/00/766200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/766200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/766200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/00/766200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/00/766200/1"}}},"commit":{"parents":[{"commit":"e1dcb79634a5dff4e01b7ab62678c65d6ced11df","subject":"Add Neutron extension check before calling some APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e1dcb79634a5dff4e01b7ab62678c65d6ced11df"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-12-08 20:28:29.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-12-09 11:00:20.000000000","tz":780},"subject":"Disable openstack-lower-constraints-jobs","message":"Disable openstack-lower-constraints-jobs\n\nThe community doesn\u0027t have a final solution, the following are what we found:\n\n* virtualenv 20.2.2 released recently with pip bumped to 20.3.1\n* pip before 20.3 is not actually satisfying the full dependency set and those\n  issues were uncovered by the newer version.\n* Disable this job for now until the community has a decent solution.\n\nChange-Id: I8cd4554a20f89df92dbad4961d5a07a21f26fae8\n(cherry picked from commit 9ea87ccf4334e893565c2fa1ee8e587f56e592a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3639befa7ef1c1e66d3c0a2960846dbe405f3b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3639befa7ef1c1e66d3c0a2960846dbe405f3b99"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
