)]}'
{"id":"openstack%2Ftrove~768524","triplet_id":"openstack%2Ftrove~master~I2684608c4e8ae71a9ded46e54cdba926af6b82f3","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I2684608c4e8ae71a9ded46e54cdba926af6b82f3","subject":"remove py37","status":"MERGED","created":"2020-12-26 04:43:20.000000000","updated":"2021-01-05 04:41:54.000000000","submitted":"2021-01-05 04:40:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768524","meta_rev_id":"e433503770616f2f330065047172e6d68c64ae63","_number":768524,"virtual_id_number":768524,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 04:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"recommended":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"all":[{"value":1,"date":"2021-01-03 23:07:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":2,"date":"2021-01-04 22:09:52.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":1,"date":"2021-01-04 22:09:52.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-26 07:19:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-03 23:07:31.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2021-01-04 22:09:52.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"9fe4b46556969b86a3781c665144f50c8443284e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-12-26 04:43:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b733fb810a38488929eeb3b294bd6e5367cac0a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 07:19:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36d2071833d44d2cb81ab2f7e64929c4 : SUCCESS in 1m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a92655dff7964e6dbfa126defdd56e62 : FAILURE in 6m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcf210afe7364df58b8bc560a7f837fa : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a537e2e397a4f5cbd5083699da17357 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98a462be50fc4c9d9b66fe262da2629c : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d87ad744e7041bbaa29a15c7b0993cb : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696b745c70e04830b6afa56eb73cbd3a : SUCCESS in 5m 20s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/095565043fbe482f829c3c0e0b5aa591 : SUCCESS in 1h 29m 32s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e31fdd0052c454889d8430445862e90 : SUCCESS in 1h 46m 49s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/67b494e0c9814816926977a916934a0e : SUCCESS in 2h 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf8a031d4f95db460002aa5c74f61f3e171a04cf","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2021-01-03 23:07:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4a84ae270e4bebcf5ad59c269c69c93891a23acf","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-04 22:09:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ffd2addbe5c9861b2616ac4072ba8bf193c8502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 22:10:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9b1c33392d31212cda7e782c63261dc045ee594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 23:57:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bb8671ac00f34a94906606a75863326c : SUCCESS in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda64ecf84214af09328a512a3a66ced : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d42d079a76d74cbca552891fbd19bb15 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31f5e75b20814c888189369224425c3b : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd4ac16444c34a56b28d4dbf0a8226ab : SUCCESS in 5m 36s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/245528f3f6b44b02a9f4da306826f33a : FAILURE in 1h 44m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e7d63dfe12340b90249ac5a71a85a202b5a328c6","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-05 01:10:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"23e71a8fbe5f92ed29bda3fe5eaaa128781b6bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 03:36:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/de6d77c5fa5e4f278b4b24e33c6ed0c6 : SUCCESS in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e380ab298674bd0aaacfd0647e51839 : FAILURE in 6m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f16751e8e34f998f0d86af482114e5 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c70f7a3b886449ec8a27c3efb404a7d5 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85ec31747471458298f85620012ca859 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa5e96fddcab406e97fe178ac164e53f : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ae6eeb44034c88bfb931c5875b7437 : SUCCESS in 5m 57s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2f2239033f2a4c148e2e4c4fceb33a5d : SUCCESS in 1h 10m 49s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc914463135e4549b66bfe81a58b03ad : TIMED_OUT in 2h 02m 30s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/90473f95c7d2494b85e4aa922f1d1b1c : SUCCESS in 2h 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f31b1a1c1d50077606e93c90820f52d2d2d6a638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 03:36:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e0b9a2b8fa8d08e6bf3e555c89897652ef113b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 04:40:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e1481a56c2a94fe8989d9b3cd28733f0 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/275b2bb0909a450284d05e3884043997 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aca3758d21f47cdbac7d3bae5f4d9de : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc699ebca4de4c6ba6a50674665e3f2f : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ade8f7f9374767b82e80ac910a853c : SUCCESS in 6m 51s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e86c9cdc838243aca8bb321c5317e12b : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":1},{"id":"456f697add2910a00b4ef5ee1046a87ae62ecd3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 04:40:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e433503770616f2f330065047172e6d68c64ae63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 04:41:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a73f049cccc4ce396f415f20dbb015c : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdb00f659dbae665dff1d3951e9972a0b867ef08","revisions":{"bdb00f659dbae665dff1d3951e9972a0b867ef08":{"kind":"REWORK","_number":1,"created":"2020-12-26 04:43:20.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/24/768524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/24/768524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/768524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/768524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/24/768524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/24/768524/1"}}},"commit":{"parents":[{"commit":"3301e2e42c8fb2e26d9db60adf001194f83ef369","subject":"Merge \"update cliff to 3.5.0 support python3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3301e2e42c8fb2e26d9db60adf001194f83ef369"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-12-26 04:40:12.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-12-26 04:40:12.000000000","tz":480},"subject":"remove py37","message":"remove py37\n\nRemove python3.7 from setup.cfg, since Wallaby\u0027s python\nsupported runtimes are python 3.6 and python 3.8[1]:\n\n[1]: https://governance.openstack.org/tc/reference/runtimes/wallaby.html\n\nChange-Id: I2684608c4e8ae71a9ded46e54cdba926af6b82f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bdb00f659dbae665dff1d3951e9972a0b867ef08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bdb00f659dbae665dff1d3951e9972a0b867ef08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
