)]}'
{"id":"openstack%2Ftrove~771582","triplet_id":"openstack%2Ftrove~master~Icd8af655b27862a29726ed8a94ba8da6c76cfb4d","project":"openstack/trove","branch":"master","topic":"configuration-group","hashtags":[],"change_id":"Icd8af655b27862a29726ed8a94ba8da6c76cfb4d","subject":"fix update parameter not take effect","status":"ABANDONED","created":"2021-01-20 09:53:50.000000000","updated":"2026-09-17 17:30:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"50d396f4cffd50ae6b24502a508ff3f264c49bbf","_number":771582,"virtual_id_number":771582,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-20 11:57:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-20 11:57:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"758d3d7e7b5d545f25fcc9a148c5004bba229700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-20 09:53:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"923f05db945370a55caa7fa25f8e72ae36c2a687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 11:57:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c350bee7e3034ef98ff9cac5f27f8622 : FAILURE in 7m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0c868aa06c4d46892bfcd928d6c728 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8627ef9c1d6447d1a3c2173ca1152da9 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49d6f3ae49ae404388a9e64665712654 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d133e7ba8be4a758716ef102e7157a2 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3085e4c339644672889c7ac9123f847d : SUCCESS in 6m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b90e32ef072d494caa931fb79ba0466d : SUCCESS in 7m 08s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5615acd97d924507b393c0ad0b3bb2d1 : FAILURE in 1h 57m 03s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eab086fea8e6464599fea8debe9ac322 : FAILURE in 1h 41m 45s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e42de98916fb4099bb62f525185b6d95 : FAILURE in 1h 41m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50d396f4cffd50ae6b24502a508ff3f264c49bbf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 17:30:42.000000000","message":"Abandoned\n\nOutdated MR cleanup.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a46f19b8959085102beffc6f79048580ec7c5cce","revisions":{"a46f19b8959085102beffc6f79048580ec7c5cce":{"kind":"REWORK","_number":1,"created":"2021-01-20 09:53:50.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/82/771582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/771582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/771582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/771582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/771582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/771582/1"}}},"commit":{"parents":[{"commit":"537a2859f6115b1fdbe1bcb968ed937fdcff5339","subject":"update ci job description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/537a2859f6115b1fdbe1bcb968ed937fdcff5339"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-20 09:36:38.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-01-20 09:49:07.000000000","tz":480},"subject":"fix update parameter not take effect","message":"fix update parameter not take effect\n\nModifying the parameters that do not need to be restarted\ndoes not take effect. When there are both need restart and\nnot need restart parameters in the configuration group\n\nStory: #2008534\nTask: #41623\n\nChange-Id: Icd8af655b27862a29726ed8a94ba8da6c76cfb4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a46f19b8959085102beffc6f79048580ec7c5cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a46f19b8959085102beffc6f79048580ec7c5cce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
