)]}'
{"id":"openstack%2Ftrove~772151","triplet_id":"openstack%2Ftrove~master~Id9f5a0e54052f5541cb0d82d208923c11dfb73eb","project":"openstack/trove","branch":"master","topic":"pdf-docs","attention_set":{},"removed_from_attention_set":{"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2026-05-09 08:58:29.000000000","reason":"Change was submitted"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2026-04-28 13:27:41.000000000","reason":"\u003cGERRIT_ACCOUNT_26285\u003e replied on the change","reason_account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-05-09 08:58:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9f5a0e54052f5541cb0d82d208923c11dfb73eb","subject":"Add PDF building","status":"MERGED","created":"2021-01-23 10:01:06.000000000","updated":"2026-05-09 08:59:33.000000000","submitted":"2026-05-09 08:58:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772151-pdf-docs","meta_rev_id":"1ce9dfa8dfd3cde99a40dcf37250550c6b899419","_number":772151,"virtual_id_number":772151,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-09 08:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-09 08:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2026-04-28 13:25:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-09 08:43:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-23 13:01:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-08 15:19:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-09-08 15:19:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"de237edd60a218901d07147cbdb5a470e9238c77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-23 10:01:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3551b874be9563d60bb0f1e554838805d2e4d949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-23 13:01:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6c94be56e180493792292ce44f0b02a6 : SUCCESS in 6m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/937e208d55d04092af023bbe241e2a0b : FAILURE in 4m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44720095bc8b49a28ff6b9264462bd77 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de367aa6795b4f71804ba5d4f7bded16 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34d1963c71d7484caff59eccae0e7f07 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f97587ccc9c4eafabbf20f672b19a09 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4299ae4e8a6c4dc7b7e44e59afa2d9eb : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb001798ccc4819ac7f4d3df11e7ab0 : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/438e649aa080430f8b9af67cce0d0ed5 : SUCCESS in 4m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e03f3ab05e664921bbdee14ce8e98c9b : SUCCESS in 5m 58s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/62e1a5c7a3c44490b7cdf290a9bc8e56 : SUCCESS in 1h 45m 53s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a8a3a10eb354984a46e36429f5f499c : SUCCESS in 1h 57m 06s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/95039fe64c3041378aae79b93b7cdf6c : SUCCESS in 2h 54m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b055c0447ac0ac7a6eebb823f78bdc8e2372d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-24 07:56:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77a15bd1ec2a20adaad2c256c83dd7bca0ad3de3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 09:11:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3f334eff0a434173a815c9921f5c3b17 : SUCCESS in 5m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b628cec0a664971822f41608c5c8543 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95331b7cf7b540418c74e08a04e51ef4 : FAILURE in 6m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170c6e66b48140dab987b236f08378c9 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8f6b76d20aa46a4949f7782a47fa636 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84bc598680fd4bbc9b15944fe775ed06 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7755d5a947344f5815c2d8c414e3589 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737050b117ab4599a6f7f6da921f0f72 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5137b6396b7e47fc9b95567f6b149052 : SUCCESS in 4m 44s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4d950448eb984d3cb696029a5e529d00 : SUCCESS in 5m 23s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/42a1b59500ff4fd795acea6048d5c910 : FAILURE in 29m 07s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df46927ed5b14184926447de5fdd1703 : FAILURE in 33m 21s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ccbf72c3275b4ac785918072aa961a9d : FAILURE in 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12631fe9dfde3cfb4521530a93b48a854debb7d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-11 02:32:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ac6504765bda60e6cc5f4430296daec2e5cf5646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 04:40:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/28ef7060f7ff4c32b576cc49400292f7 : SUCCESS in 6m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cde69fe41719484492fe55cae8efa080 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6c3ccbc2beb4ef8be3026aff03f3647 : FAILURE in 7m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff04d6145c6466c8236b3a1e1c8c1d7 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8924b642d3047b2992767dacef59ab2 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4049edd122f94cb1b3a4b751114ed189 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b69185842bf84dd4b503a208db45caa0 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64130486acd64a8eb2b784fd2a337ac3 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c0a8f7b22a34da883f85d6fe223be6c : SUCCESS in 5m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f689adf0b26495a8ff6c12ab82625b5 : SUCCESS in 5m 42s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5d2afa89534b4ae5aa32b0e42f95b35a : SUCCESS in 1h 27m 30s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/2d2c4a90b7be40f1aebc98e5bd5d0790 : SUCCESS in 1h 34m 13s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adf1682d0fb24b9d9326981740c77915 : TIMED_OUT in 2h 03m 44s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/54e3011624874f848905dca35118c2b4 : FAILURE in 45m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"611a64009f6a87153116df77f92de962ae24aad4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:46:58.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7e4d60ffc88679f584680f7a09f6b692834d0d8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:47:18.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"8851db37a8271f80993b9fed62d161485f8d7d96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:48:41.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"7170765cf08b711b77fc466f2f890142b2a36306","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:49:46.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"c474cdf37af111a898a6c6b750ed6e73b7fad6f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:49:50.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"7dc455119aa3510a26010791e9fb592b86c6c649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 12:52:10.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"35ddff41ec5e10b4449c7426be998ad1290df23c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:07:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2d88f6b5384eb9b2646177ebd9ef0f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aee638fc34764ba8bf274a8c8641c545 : SUCCESS in 5m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d68586d7be364c81b631889a8aae4df5 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/923944f02a534104a1da655db4413166 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fecafca4ba5a4de38c9b0f6fce06addb : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/36dd5ab52a384ff89c285a37514c7eb3 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9a1d5825b8b4b478a03ecb29b4ff7b2 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23318d635a004ce3be75a3166b274ce3 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7393d16df14407f992355d0abd88b9e : SUCCESS in 5m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/69fd000add004685be165a48a7a82fce : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":9},{"id":"59a8182e5a55d815be8dbd15d79a7ed804a163fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 13:32:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/759acd9e2fd84a5e979d4eb28d28d7fa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5eb9b4cefc5b4497a57b85819df3e868 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9cc7f9aa42794fdeb9df4e29fbc529eb : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c17f93323d04338a576bf0ba44ada6f : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0b75858fa30efe610ddf866af6bc5e671a95bd1c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-08 15:19:53.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"4e34cc9095316a46e873b9c33a95442ff7b46524","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-04-28 11:49:34.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"bab5192e7fcefc448f6cee27e806816a095a1a0f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-04-28 13:02:18.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2574dedbb03ad2b727faaed32dc15edf559bbb83","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 13:07:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"43cac7bfa28bfb78360fc80471db4a28e2d3d7cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-04-28 13:19:51.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"26d1035958807ef31e3d0edb53f5d1b58bfaf6f7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-04-28 13:23:35.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fdc98f8690e1dca305eb9d8565878f8b43f1571a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-28 13:25:00.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d4f031e1f275c56ecf15bb7780d211c68786c8e7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-04-28 13:27:41.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"468833db9314dd9892331670916e010672269274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 13:35:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f97d88a4efc49149cf3ec79e93816d8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/21b9ca7321bc4b25b260955e2f108b15 : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c59adf5b056945d2896a839f9c6fa78c : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c44a127d70174d3c8f4d1892fd82acc7 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a214e572d6e24b19b4c8fe6c1d36764a : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fe2126c583840e3a3e17b9f3c17076b : SUCCESS in 6m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3dc5f82a95fa41a09b68ad736dd868f3 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9abc5b32e7e4731bbdc587ec604bbf4 : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17eb329618d646ecaa9279f3aee4f08d : SUCCESS in 4m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/235dc6be640a48759871602d9545a4ab : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":10},{"id":"8ea9ce9ff3d05a942534614d00fb2882b192264f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 13:54:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dfa98bbfb8648d7b3cd8e5bb03d1245\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5f93c77f02e415d8a43fe6b01f851b2 : SUCCESS in 11m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da21ef36fa1c46a28a8804ef83542ad0 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/95ad69fda5204abdbbdefe8987c787a3 : SUCCESS in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"663c1f7e93d4636bb221e20cf566b232dfe6e8b6","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2026-05-09 08:43:48.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"8146130c874f93883170d2625c5ac63de5ecabb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 08:44:00.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"1be61d8f73baebce00e2d38672560ae884f4088c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 08:58:29.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3305d4055b824a0795661dafc8c7ae61\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/836459170f414df7b9bfcd6742306167 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/922b14e83cfe41f193a8030d6959e91b : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6adb9d0a1f64138a2bff567238a3fee : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07b23119fcdf493181049ff26fe2116d : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae99e0ebfb1473f87c971a4a100f02f : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58ad27ecee7c4786a4b50e0f0d932577 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":10},{"id":"7ebc8dc148619ed84969c315efb021fd19f7c2c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 08:58:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"1ce9dfa8dfd3cde99a40dcf37250550c6b899419","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 08:59:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29fe2bb8597744649024bbd9a3ce808c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/345e3cacb46c469c89bfc64328f962c4 : SUCCESS in 45s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/411fcc8befce4ce9a272a6c528ad0d02 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2be558b127ce471595d959babb012f3c : SUCCESS in 50s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"4a57782d89b37d87800ed1eda4bec72a6d759bac","revisions":{"2b0b690621825ba0479fa0be5033530fbc44d190":{"kind":"REWORK","_number":1,"created":"2021-01-23 10:01:06.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/51/772151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/1"}}},"commit":{"parents":[{"commit":"537a2859f6115b1fdbe1bcb968ed937fdcff5339","subject":"update ci job description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/537a2859f6115b1fdbe1bcb968ed937fdcff5339"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:53:58.000000000","tz":480},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b0b690621825ba0479fa0be5033530fbc44d190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2b0b690621825ba0479fa0be5033530fbc44d190"}]},"branch":"refs/heads/master"},"0c48be608266b25668c500ac6335c9b2b34f53a6":{"kind":"REWORK","_number":2,"created":"2021-01-24 07:56:36.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/51/772151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/2"}}},"commit":{"parents":[{"commit":"537a2859f6115b1fdbe1bcb968ed937fdcff5339","subject":"update ci job description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/537a2859f6115b1fdbe1bcb968ed937fdcff5339"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-01-24 06:15:13.000000000","tz":480},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0c48be608266b25668c500ac6335c9b2b34f53a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0c48be608266b25668c500ac6335c9b2b34f53a6"}]},"branch":"refs/heads/master"},"1b97191b7401228952f5bde9cd1b4e596b9a7f0b":{"kind":"REWORK","_number":3,"created":"2021-05-11 02:32:05.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/51/772151/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/3"}}},"commit":{"parents":[{"commit":"71d7b183193b8aa94b1eae4831a108b8c3534c30","subject":"Merge \"setup.cfg: add oslo.policy.enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d7b183193b8aa94b1eae4831a108b8c3534c30"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 02:31:26.000000000","tz":480},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b97191b7401228952f5bde9cd1b4e596b9a7f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1b97191b7401228952f5bde9cd1b4e596b9a7f0b"}]},"branch":"refs/heads/master"},"6e1f723f6d626710ce173c215bb1a052befb1173":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-09-08 12:46:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/772151/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/4"}}},"commit":{"parents":[{"commit":"71d7b183193b8aa94b1eae4831a108b8c3534c30","subject":"Merge \"setup.cfg: add oslo.policy.enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d7b183193b8aa94b1eae4831a108b8c3534c30"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:46:58.000000000","tz":0},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\nSigned-off-by: likui \u003clikui@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e1f723f6d626710ce173c215bb1a052befb1173"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e1f723f6d626710ce173c215bb1a052befb1173"}]},"branch":"refs/heads/master","description":"Edit commit message"},"34368f49c69d860daa279cb4927617e02097630c":{"kind":"REWORK","_number":5,"created":"2025-09-08 12:47:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/772151/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/5"}}},"commit":{"parents":[{"commit":"71d7b183193b8aa94b1eae4831a108b8c3534c30","subject":"Merge \"setup.cfg: add oslo.policy.enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d7b183193b8aa94b1eae4831a108b8c3534c30"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:47:14.000000000","tz":0},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\nSigned-off-by: likui \u003clikui@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/34368f49c69d860daa279cb4927617e02097630c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/34368f49c69d860daa279cb4927617e02097630c"}]},"branch":"refs/heads/master"},"5a0177e55bf6644fcec46e6a952453d6b8c18985":{"kind":"REWORK","_number":6,"created":"2025-09-08 12:48:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/772151/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/6"}}},"commit":{"parents":[{"commit":"71d7b183193b8aa94b1eae4831a108b8c3534c30","subject":"Merge \"setup.cfg: add oslo.policy.enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d7b183193b8aa94b1eae4831a108b8c3534c30"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:48:41.000000000","tz":0},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\nSigned-off-by: likui \u003clikui@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a0177e55bf6644fcec46e6a952453d6b8c18985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5a0177e55bf6644fcec46e6a952453d6b8c18985"}]},"branch":"refs/heads/master"},"fece138f105e3357fe8a6ad7de539679eee63938":{"kind":"REWORK","_number":7,"created":"2025-09-08 12:49:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/772151/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/7"}}},"commit":{"parents":[{"commit":"71d7b183193b8aa94b1eae4831a108b8c3534c30","subject":"Merge \"setup.cfg: add oslo.policy.enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d7b183193b8aa94b1eae4831a108b8c3534c30"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:49:43.000000000","tz":0},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\nSigned-off-by: likui \u003clikui@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fece138f105e3357fe8a6ad7de539679eee63938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fece138f105e3357fe8a6ad7de539679eee63938"}]},"branch":"refs/heads/master"},"005b38f090c34ca01d43fced5390ec50da1d93d4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-09-08 12:49:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/772151/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/8"}}},"commit":{"parents":[{"commit":"4beb14657745bac15c745e955936e6dbd8f42e63","subject":"Remove pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4beb14657745bac15c745e955936e6dbd8f42e63"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:49:50.000000000","tz":0},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\nSigned-off-by: likui \u003clikui@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/005b38f090c34ca01d43fced5390ec50da1d93d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/005b38f090c34ca01d43fced5390ec50da1d93d4"}]},"branch":"refs/heads/master","description":"Rebase"},"52df5ac5a58356b25cc4f68955c5543a3cd267f9":{"kind":"REWORK","_number":9,"created":"2025-09-08 12:52:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/51/772151/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/9"}}},"commit":{"parents":[{"commit":"4beb14657745bac15c745e955936e6dbd8f42e63","subject":"Remove pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4beb14657745bac15c745e955936e6dbd8f42e63"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-08 12:52:10.000000000","tz":0},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\nSigned-off-by: likui \u003clikui@yovole.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52df5ac5a58356b25cc4f68955c5543a3cd267f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/52df5ac5a58356b25cc4f68955c5543a3cd267f9"}]},"branch":"refs/heads/master"},"4a57782d89b37d87800ed1eda4bec72a6d759bac":{"kind":"REWORK","_number":10,"created":"2026-04-28 13:19:51.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/51/772151/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/772151/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/772151/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/772151/10"}}},"commit":{"parents":[{"commit":"573cef64893ba7f8b0f777b49c26c9ce206c186e","subject":"Revert \"Fix backup failure when caller uses an application credential\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/573cef64893ba7f8b0f777b49c26c9ce206c186e"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-01-23 09:49:57.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wchy1001@gmail.com","date":"2026-04-28 13:19:41.000000000","tz":480},"subject":"Add PDF building","message":"Add PDF building\n\nThis commit adds PDF documentation build target \u0027pdf-docs\u0027 that will\nbuild PDF versions of our docs. As per the Train community goal:\n\nhttps://governance.openstack.org/tc/goals/train/pdf-doc-generation.html\n\nChange-Id: Id9f5a0e54052f5541cb0d82d208923c11dfb73eb\nSigned-off-by: likui \u003clikui@yovole.com\u003e\nSigned-off-by: wu.chunyang \u003cwchy1001@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4a57782d89b37d87800ed1eda4bec72a6d759bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4a57782d89b37d87800ed1eda4bec72a6d759bac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
