)]}'
{"id":"openstack%2Ftrove~772610","triplet_id":"openstack%2Ftrove~master~Ie704db8233c4b797eeefa140c4a11427280dffd9","project":"openstack/trove","branch":"master","topic":"fix-python2","hashtags":[],"change_id":"Ie704db8233c4b797eeefa140c4a11427280dffd9","subject":"Stop using pip-and-virtualenv diskimage element","status":"MERGED","created":"2021-01-26 23:14:09.000000000","updated":"2021-01-27 07:21:05.000000000","submitted":"2021-01-27 07:18:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772610-fix-python2","meta_rev_id":"b9ad2b47050f997bca94bbd8d2542042880add38","_number":772610,"virtual_id_number":772610,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-27 07:18:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-01-27 05:35:14.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-01-27 05:35:14.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 01:30:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81ad34b0875a0aa1e7e80caa8955ca0734e87848","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-26 23:14:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a2f6a6a05b3dd11166e314241124403f298404f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 01:30:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44839e02e7bb4c4294b16841b1bd3a53 : FAILURE in 7m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed2ea43e104e4f009f1c98a0c1271f95 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7444dfb9e7c04de09ca02832cd0ea414 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/249b5787739d4114a5899681d2184fa9 : SUCCESS in 7m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e0d2cfcfe434ad6ba6ff14cc7ba08ee : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6a9e8e0b9d4922a492075af510a3c5 : SUCCESS in 5m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6b855bd03be7477ea90567803d5132e1 : SUCCESS in 5m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a55c0439fe764379905f57adbc2cb2a0 : SUCCESS in 1h 17m 13s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c6a9e8453ae47c4b4f4fa194470efe6 : FAILURE in 1h 38m 06s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e60cac8f92fa4374a859bcf33f2b8d57 : SUCCESS in 2h 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d03e9ff52e34eff40339ec31afe1dfd7af1ea4be","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-01-27 05:35:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"400b1a88e58a17b44e8186e7516d9f515461d5e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 05:35:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f85f9f1c9435ac46fed7635471b09d6345b02a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 07:18:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41f2a695e05844f0906f0ef183a1cdf9 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1128a293b38b4202b382213fd14fffa6 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31057f5e0a754067a79e0b00088bd35e : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b608859785f647a2b7af8fb236bfffa7 : SUCCESS in 5m 57s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/85b5fb0920934d3e838e7a30807031a2 : SUCCESS in 1h 42m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4acce1772a798e0aae549aa62cd1b98e37d19700","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 07:18:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b9ad2b47050f997bca94bbd8d2542042880add38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 07:21:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/880e428065b4483eb265d193fd223333 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e04b269ca75067e28ae3e6ecb60ac2d11ab5b3b","revisions":{"1e04b269ca75067e28ae3e6ecb60ac2d11ab5b3b":{"kind":"REWORK","_number":1,"created":"2021-01-26 23:14:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/10/772610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/10/772610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/772610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/772610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/10/772610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/10/772610/1"}}},"commit":{"parents":[{"commit":"537a2859f6115b1fdbe1bcb968ed937fdcff5339","subject":"update ci job description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/537a2859f6115b1fdbe1bcb968ed937fdcff5339"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-26 21:38:02.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-01-26 22:43:16.000000000","tz":780},"subject":"Stop using pip-and-virtualenv diskimage element","message":"Stop using pip-and-virtualenv diskimage element\n\nALthough diskimage-builder will solve the issue eventually, but we\nstill prefer to put the package installation under trove\u0027s own control.\n\nStory: 2008559\nTask: 41697\n\nChange-Id: Ie704db8233c4b797eeefa140c4a11427280dffd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1e04b269ca75067e28ae3e6ecb60ac2d11ab5b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1e04b269ca75067e28ae3e6ecb60ac2d11ab5b3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
