)]}'
{"id":"openstack%2Ftrove~774762","triplet_id":"openstack%2Ftrove~master~I98baf252452353237bc8fb14357df4e7bcb2867a","project":"openstack/trove","branch":"master","topic":"fix-postgresql-enable-log","hashtags":[],"change_id":"I98baf252452353237bc8fb14357df4e7bcb2867a","subject":"Do not override RESTART_REQUIRED service status","status":"MERGED","created":"2021-02-10 02:53:15.000000000","updated":"2021-02-10 09:57:35.000000000","submitted":"2021-02-10 09:55:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774762-fix-postgresql-enable-log","meta_rev_id":"6d46c55f3e26124af94138da4b47085f717990ed","_number":774762,"virtual_id_number":774762,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-10 09:55:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-02-10 07:59:58.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-02-10 07:59:58.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 05:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8290fc377e285143e3c2ffdeb610e417d269e6bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-10 02:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c232c4b31ab8b852d72b0d77d8b3d146ca23d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 05:48:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3db4ed87d3384e0aacf4fd2921e25049 : FAILURE in 5m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d2b820028e4bb69bd706c2cc2784c0 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc34a068dbd141f494dd9bea12a640e7 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ad2b41aecca4961b80f438e806d581e : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b043974795a4b73a2cc6a5b24257610 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c4b77a023a447fbff795a0b21cd68c : SUCCESS in 5m 39s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/af755e7237844a2eb41a5a8358cd848e : SUCCESS in 5m 17s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a063436bd95140b5b2575aa5d4d9aff9 : SUCCESS in 1h 33m 51s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39f30d55cd4a4a1abff011552ca10fe6 : SUCCESS in 1h 44m 44s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0e4b43ed8bb646889e52b8d24d66ab79 : SUCCESS in 2h 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7bfd579a4d40b4cc7b4e7a0574642a504137357","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-10 07:59:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"851e21b6cf2e5a2db7b9189135349e3829cb1b6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 08:00:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46b451eeec858077f69c4042b7cf5d205fa76552","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 09:55:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c2379c6ab04e87bbad7ff1c4ddfabe : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7ce3db28bd740cc888380d55406352a : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e000001c267d409984fc0151f49d7b84 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3587da78b0041e5af6bc7128dc0c98b : SUCCESS in 5m 30s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6d3eee981cd24e1dab7c271abdb4939d : SUCCESS in 1h 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"89272c58c65a67f09cb8e7abfcb902c2afb69b65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 09:55:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6d46c55f3e26124af94138da4b47085f717990ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 09:57:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ae079cd4a34e0dbf2f1e8c7b8a8dcb : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","revisions":{"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5":{"kind":"REWORK","_number":1,"created":"2021-02-10 02:53:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/62/774762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/62/774762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/774762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/774762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/774762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/62/774762/1"}}},"commit":{"parents":[{"commit":"6edd3332b68864e69eb6733a71ecb5e827871d8f","subject":"Doc: custom container image registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6edd3332b68864e69eb6733a71ecb5e827871d8f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-09 20:53:06.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-10 02:51:54.000000000","tz":780},"subject":"Do not override RESTART_REQUIRED service status","message":"Do not override RESTART_REQUIRED service status\n\nThe RESTART_REQUIRED service status set by guest agent should not\nbe overridden by the health heartbeat mechanism.\n\nThe RESTART_REQUIRED can only be changed when the instance is restarted\nor rebooted.\n\nStory: 2008612\nTask: 41795\nChange-Id: I98baf252452353237bc8fb14357df4e7bcb2867a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
