)]}'
{"id":"openstack%2Ftrove~775529","triplet_id":"openstack%2Ftrove~master~Ie370eff134ae0e1394d0b03049a46a125d0db71a","project":"openstack/trove","branch":"master","topic":"fix-replication-log","hashtags":[],"change_id":"Ie370eff134ae0e1394d0b03049a46a125d0db71a","subject":"Remove the incorrect log for creating replication cluster","status":"MERGED","created":"2021-02-14 20:07:36.000000000","updated":"2021-02-18 23:24:53.000000000","submitted":"2021-02-18 23:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775529-fix-replication-log","meta_rev_id":"f40b647c28036a4775629e5e8f44f9dc7845e318","_number":775529,"virtual_id_number":775529,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-18 23:22:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-02-18 21:38:49.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-02-18 21:38:49.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-14 20:41:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1cf6059f7ad8bb10725aff9806c1ba955e796caa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-14 20:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"118d738ab5e94290e16a43cb37cc97158392b36e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 20:41:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06b4c51b6f424c499f50326d0c8d6e9f : FAILURE in 6m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b5fb6489f7249d095c06e1ca833b35a : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5462f712eaa44be890c10e1c33bbdaad : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09e65902d87f4aaf92791c517ce965ef : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f97eead66af64c28a031bb6ad3e10ee0 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b70a30fccff482f9900cbadf0096e40 : SUCCESS in 5m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c81d442264ad483fb613aad9e44c5db3 : SUCCESS in 5m 17s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/062e01f8601c4bc0b120f84d4433c22a : FAILURE in 19m 10s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41d4ccf56a514c67b02826f8674d456f : FAILURE in 26m 09s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e097766716f340b18fd1e59ed9912729 : FAILURE in 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0248887fbe840d141a11dd67e8309fe3723e3bc6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-15 19:33:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4bac997ea654ee06601fd78d269d7af743f707ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 20:10:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12dfaa86b509491b9f8558c4fca19c91 : FAILURE in 12m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/661c0f1a39ef4c709582f52ea5e26e73 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d30092ff5fb4918a8122f6de106f422 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/968ef012e06a4f418f16cfd6bee4976a : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f707c2d5a634bbebdabdf0b8e8726c2 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9522c7a6dfca4cc985f32bce4b82f40c : SUCCESS in 5m 49s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac84e06eb85a4d2fa6256d4cfd519346 : SUCCESS in 5m 45s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4cf4248dec234b91a0270d15ba94208d : FAILURE in 22m 41s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2404af6ad1054a45828fe030dfcb9cd9 : FAILURE in 27m 47s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ff04f63bebc74b0aacad8cd6791220fd : FAILURE in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee8dea592cb78a277fc00cffef945b391cc390b4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-16 20:26:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41b92e9c5dab805b7292369b83fbb3359fc4a4c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 20:59:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad89caef8c414bcdbc48aad826ac2858 : FAILURE in 8m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bcf67805110469abe89bb877ccad0a6 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3be4e595025d4d48af76823e80cabc45 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c4b7bb421a445b2ac35391d8c2ce851 : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89e5ba3076d8421585479f615298a783 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7119182dcc4444c9cffadb66accafc5 : SUCCESS in 6m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1804f7d986f540d4a43092cc7872c9ab : SUCCESS in 5m 26s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bdf0e9b5e20b43e8b75e40f2697ec4e8 : FAILURE in 22m 08s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eedc93aa9314759a373f167d618d32c : FAILURE in 22m 31s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/957e6b5f6e9d41a9a688deb61bf2489b : FAILURE in 22m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"235947d36695178e45feb4df93623cdecf7850c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-16 22:17:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"970e0003256845b1dacc24ed51bf71f20c78c260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 22:36:08.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"c63b67bece06ba0e871d131454d21a4fa567bb41","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-16 22:48:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1dc52fccd7a524b03217b20275a4109e7f0d7849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:14:17.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"50425815762c798dfc79da9d3ddc51a392c2ffed","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-17 02:52:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dd259bca3020582c16828b89aba1ec0f529afd91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 05:45:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/483fdc7f3fe94dcdb0f2fc925575c8ab : FAILURE in 6m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c81e00d7a9494899ff095c38ae6311 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8531d76b881b41f2aa40a0238fdffb86 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e342f8b780d84c9da4e475989076ea96 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/763980917ae0457bbeb8b8244b1527c3 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c1759a101144dffba2ec6fdc3bc60b8 : SUCCESS in 8m 00s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0cc31c83421b4911adf8fa8e54090d0f : SUCCESS in 7m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9d3c09fd597d4632b507b0a8db62cc81 : SUCCESS in 1h 44m 48s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c96138fdc834537adc4370c99f9c137 : TIMED_OUT in 2h 03m 23s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ad7d3aa6671d406786006b0e8d23c8e1 : SUCCESS in 2h 45m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d8420c597419501d713a15a02ed772f5394e69d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-17 09:24:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2335083a8fd54bcdc02c40dbd89978ff0e1db788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-18 17:33:50.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f10dbf0dd9cd1877fce45c64dbe45946f2fcb02b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 20:42:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b1eba15cc864183abed735b78b51f9c : FAILURE in 7m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d28a98ec5a34971a07f3ee553706fb3 : SUCCESS in 9m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da71fb3129234362a49bcebd24e8c755 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56936cbaa3b049afaf6f84e37dbbda5b : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/917576b4cee84eacacb3f82906f575ab : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7fede35e724bbfb74f3ed250175184 : SUCCESS in 7m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b151c39c0cab4d3883260c685226573f : SUCCESS in 6m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/63f1c02df893481d975e51e49f43f9b9 : SUCCESS in 2h 12m 08s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8d380a1b27d430a99743725a7f44bd0 : TIMED_OUT in 2h 03m 59s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/13bea3218b4b4af184db2ad03a98164c : SUCCESS in 2h 54m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"024fafe19b4ac884f9ff3029e068afec1b953ea4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-18 21:38:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cacf4a779f11856716cef11d7cfd09e50e13c9f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 21:39:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f651b82285241f668bd66c59a44048a1dad7a620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:22:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3bc80d84a3f4280b4bc77981d5c113c : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3efcfbac67a4adb867e36fdb2fc2635 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01a0e5b4c1c54ae3a93ac24224873dc5 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2c18be43a6245f3b6ba5cd81eed3c43 : SUCCESS in 7m 26s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8bd68537c46c4e3fae7b93c462a3022d : SUCCESS in 1h 41m 40s","accounts_in_message":[],"_revision_number":3},{"id":"2a84ac4b55ab608e0971f494da0cf007573087b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:22:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f40b647c28036a4775629e5e8f44f9dc7845e318","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:24:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6beb98bb99cc43adad74a4ba01aa0561 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f7fa5518afca91aa4a2a3bfd3efaff51397f108","revisions":{"bc1d7e97f2d53cd6077dbaa4a9a0edfa7092ef89":{"kind":"REWORK","_number":1,"created":"2021-02-14 20:07:36.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/29/775529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/29/775529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/775529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/775529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/775529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/29/775529/1"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-14 20:06:35.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-14 20:06:35.000000000","tz":780},"subject":"Remove the incorrect log for creating replication cluster","message":"Remove the incorrect log for creating replication cluster\n\nChange-Id: Ie370eff134ae0e1394d0b03049a46a125d0db71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bc1d7e97f2d53cd6077dbaa4a9a0edfa7092ef89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/bc1d7e97f2d53cd6077dbaa4a9a0edfa7092ef89"}]},"branch":"refs/heads/master"},"0d1bfb0fcdfc312133c07c5d9143cfc66188f598":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-16 22:17:48.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/29/775529/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/29/775529/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/775529/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/775529/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/775529/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/29/775529/2"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-14 20:06:35.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-16 22:17:40.000000000","tz":0},"subject":"Remove the incorrect log for creating replication cluster","message":"Remove the incorrect log for creating replication cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/775940\nChange-Id: Ie370eff134ae0e1394d0b03049a46a125d0db71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0d1bfb0fcdfc312133c07c5d9143cfc66188f598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0d1bfb0fcdfc312133c07c5d9143cfc66188f598"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9f7fa5518afca91aa4a2a3bfd3efaff51397f108":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-18 17:33:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/29/775529/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/29/775529/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/775529/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/775529/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/29/775529/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/29/775529/3"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-14 20:06:35.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-18 17:33:46.000000000","tz":0},"subject":"Remove the incorrect log for creating replication cluster","message":"Remove the incorrect log for creating replication cluster\n\nChange-Id: Ie370eff134ae0e1394d0b03049a46a125d0db71a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f7fa5518afca91aa4a2a3bfd3efaff51397f108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9f7fa5518afca91aa4a2a3bfd3efaff51397f108"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
