)]}'
{"id":"openstack%2Ftrove~775532","triplet_id":"openstack%2Ftrove~master~I348a82aa948079b7bba5c172b8d57d5f5dee28a0","project":"openstack/trove","branch":"master","topic":"change-agent_heartbeat_expiry","hashtags":[],"change_id":"I348a82aa948079b7bba5c172b8d57d5f5dee28a0","subject":"Change the default agent_heartbeat_expiry to 90s","status":"MERGED","created":"2021-02-14 21:30:21.000000000","updated":"2021-02-18 23:22:45.000000000","submitted":"2021-02-18 23:22:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775532-change-agent_heartbeat_expiry","meta_rev_id":"42752893de6f1899417c16c840e6542a56829f4a","_number":775532,"virtual_id_number":775532,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-18 23:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-18 21:39:08.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-18 21:39:08.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-14 21:56:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05cc6a71ccba5d17c9e492f94d86b8dc3c75abeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-14 21:30:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d1b94830e549051fafe094b60f9d527715d4d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 21:56:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3b727d6a313426bb277cb64f0f3b599 : FAILURE in 8m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7294f7229de84ec2a76201d457f7f977 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a062aab301c4d84b65c19e4f1fb3352 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/122d8970c7954be18f890fb34ba7f60e : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56726d1cc8034d838e2cebfc328bbe2d : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb31b8fd29d4436aaecc8f4e69e5ddb : SUCCESS in 8m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c217d639693740f1ad536d86d843f4ca : SUCCESS in 5m 05s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e293d92eba994f93a872499a2d805c37 : FAILURE in 23m 32s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e6790dd36ad4744a52f7d12037c632e : FAILURE in 21m 07s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/97ea9a54f3a2448685e3470eebf7942b : FAILURE in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c2b2bd26a7dc049630c3e188e626719edd3cc2a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-15 19:34:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e731b1b424817c76c0d33235635aeda161959f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 20:13:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4a5e3d0ea8b47a5bad5d76ebbac8438 : FAILURE in 6m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6480879d7bac4484b7383ecb2d1fdc39 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff8c60252756450bb1ad582713d1577e : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56fc807ba17b4086903a5e440c722c7b : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c29891b86ca401a87d9976404d0faf1 : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8a40a6ce6964cb896829a8891671bc3 : SUCCESS in 5m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ade2d965e15d4c189ace6ffdde0ad6dc : SUCCESS in 7m 42s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0d32046beea34f41a4faf9ef39829525 : FAILURE in 24m 30s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41f442a12ef54a5ea232ad7241bd11df : FAILURE in 26m 15s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2af65d266970444a88ec03bbf8784535 : FAILURE in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3587266e83830bff66a6df4eb8c7a546f365cba6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-17 09:25:54.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6f617a29d31ca8a6eace46da28d7479e1b54c074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 12:32:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dd802c0df0e4dd58ef4d8a55d52a28c : FAILURE in 7m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ed6ea100a048ef9f35d19f026fa82d : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa9c47f6cad84b5380eba273909ff240 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34301d7448ad448e861c8a4708f6cfc1 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c703ff843754714a6e5a1ac511ea1b6 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68eff5d1b3954bce8e68d18265f70d67 : SUCCESS in 8m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/cd7dcbcbe03f465ba70a62055c29ef66 : SUCCESS in 6m 33s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3207131eff7f4ae18a0469bf4e43f67d : SUCCESS in 2h 00m 47s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c7e8a8203fe4b998bbaa9602cf11218 : SUCCESS in 1h 55m 05s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/bef64c1a639e400b8971106ca338a4ab : TIMED_OUT in 3h 03m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"798229f380a3cb19b841ef4e60c9978f59f36faa","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-17 19:28:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b6c9b0184ff2a600ffa7fffbc6d1f7512aae169e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-18 17:34:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0247c84c32bf02242d03b39e1d7e2a5787aaabdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 20:43:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22a8d05f1a124ae48a9ac85920e5d706 : FAILURE in 7m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50eb157f70d94cd8877601f2952f0a39 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/842b159f30b445099b5d1d20746033ea : SUCCESS in 9m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65cd20677d0b448d8b9805bb10314bd9 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae9ec79da22a4e20b2397f70c9f0293b : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db40f9977ae149babb68e8e048ccb9ee : SUCCESS in 7m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8575d493690e4fa29bc4d3672fb00f2d : SUCCESS in 7m 13s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/665226eaff164e4e8940df9543c428cc : SUCCESS in 1h 33m 42s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd40da83929347f1b4b3b4fc73f78317 : SUCCESS in 1h 46m 09s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/004c1934a98d4f3aa5672f2174b3d5d7 : TIMED_OUT in 3h 05m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d18b703bf5ac4b483e7dac945ec050f9b262636f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-18 21:39:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"064468fd3b8c1fd2b5b3394df5c347afbf19d52c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 21:39:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"faebb32f075e18e038118b133fea0807b017b9eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:22:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73dfef552d8f4b8595e76e7cff9fa280 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80ac8dba91774bc6a2ab98e89f768397 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61be5b9b5e0242a1ac4b53f905bb64d3 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/280ca2fcfaf549c2809255262cdd5c30 : SUCCESS in 6m 44s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d11dd113bd9d49269295d45c2fcc1c41 : SUCCESS in 1h 38m 20s","accounts_in_message":[],"_revision_number":3},{"id":"42752893de6f1899417c16c840e6542a56829f4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:22:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5640a61acbfd3449ba700f495aa885f54fd2e080","revisions":{"56dd06802f075d5a271c85576500d5b23e0eac04":{"kind":"REWORK","_number":1,"created":"2021-02-14 21:30:21.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/32/775532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/775532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/775532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/775532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/775532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/775532/1"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-14 21:29:24.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-14 21:29:24.000000000","tz":780},"subject":"Change the default agent_heartbeat_expiry to 90s","message":"Change the default agent_heartbeat_expiry to 90s\n\nBy default, trove guest agent is sending health heartbeat msg for every\n60s, so agent_heartbeat_expiry should be bigger than that.\n\nChange-Id: I348a82aa948079b7bba5c172b8d57d5f5dee28a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56dd06802f075d5a271c85576500d5b23e0eac04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56dd06802f075d5a271c85576500d5b23e0eac04"}]},"branch":"refs/heads/master"},"6b42c63bc32dde751d8368b1694bb16fb6abed55":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-17 09:25:54.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/32/775532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/775532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/775532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/775532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/775532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/775532/2"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-14 21:29:24.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-17 09:25:50.000000000","tz":0},"subject":"Change the default agent_heartbeat_expiry to 90s","message":"Change the default agent_heartbeat_expiry to 90s\n\nBy default, trove guest agent is sending health heartbeat msg for every\n60s, so agent_heartbeat_expiry should be bigger than that.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/775940\nChange-Id: I348a82aa948079b7bba5c172b8d57d5f5dee28a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b42c63bc32dde751d8368b1694bb16fb6abed55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6b42c63bc32dde751d8368b1694bb16fb6abed55"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5640a61acbfd3449ba700f495aa885f54fd2e080":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-18 17:34:05.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/32/775532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/32/775532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/775532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/775532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/32/775532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/32/775532/3"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-14 21:29:24.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-18 17:34:01.000000000","tz":0},"subject":"Change the default agent_heartbeat_expiry to 90s","message":"Change the default agent_heartbeat_expiry to 90s\n\nBy default, trove guest agent is sending health heartbeat msg for every\n60s, so agent_heartbeat_expiry should be bigger than that.\n\nChange-Id: I348a82aa948079b7bba5c172b8d57d5f5dee28a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5640a61acbfd3449ba700f495aa885f54fd2e080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5640a61acbfd3449ba700f495aa885f54fd2e080"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
