)]}'
{"id":"openstack%2Ftrove~775928","triplet_id":"openstack%2Ftrove~master~Iefef3bf969163af707935445bc23299400dc88c3","project":"openstack/trove","branch":"master","topic":"restore-backup","hashtags":[],"change_id":"Iefef3bf969163af707935445bc23299400dc88c3","subject":"Support to restore backup from remote location","status":"MERGED","created":"2021-02-16 20:20:48.000000000","updated":"2021-02-18 23:26:31.000000000","submitted":"2021-02-18 23:22:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775928-restore-backup","meta_rev_id":"9aedcf99648fa9ae3fa24799ca34129397584469","_number":775928,"virtual_id_number":775928,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-18 23:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-02-18 21:39:26.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-02-18 21:39:26.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 20:59:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc9b40cec27089f2da6a29f1e0456a658d82b9f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-16 20:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c506692e024a2e73931fe2c824aecfa7992cfa01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 20:59:00.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ec6adb58b878482db6a540176ffecc7e : FAILURE in 6m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03ad754ef5984ae5876aeab913577bac : FAILURE in 6m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4460cfedc85c41b08cb85161f83b2358 : FAILURE in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/869ae1e2494d44ad8b82d5c054c270ab : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c7a9f4bd9f54a9d98e953aebbc330df : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a624ec4820e94534a6f434357e67f947 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6618a24600e4f88a5ee0e1fa5f9552c : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/195c1f38c9c149838e74e3a7a98e21ed : SUCCESS in 5m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5687eb68034646b7b093f330070897d0 : SUCCESS in 5m 43s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8014c18618e14f3bb65181325fa8e9f1 : FAILURE in 33m 10s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a86bc3b6a97b4cad80abc34b7646456e : FAILURE in 26m 50s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/279f027665624bd0b331fe708e56582d : FAILURE in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51fa72d7d9b34a2f47490624f540cd6e7552f50b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-17 09:26:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c9f8a7e47320b27f91052f658e30a15d3e6f366a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 12:23:16.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/208555a31bd44fb48d5598d5b00f37a3 : FAILURE in 6m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48311f234bef42a6a69c2c1b55e0a063 : FAILURE in 7m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a2d439b3c8a4a139371897b6b85b040 : FAILURE in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c61fb500d80495f97c8b292d9b6f744 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bb8496c77a64b35a05d0f87f08be46e : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09715b313c5c43d388fb04f4e2b9cf54 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f716f9b0adb487686f3e32218f0e018 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c2eb0bc705f4f539ba2a0fe268f5846 : SUCCESS in 6m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c0edf75df8bd4f309c84e93305451990 : SUCCESS in 6m 14s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3aa76b8316494e4ead26e11e701bfc49 : SUCCESS in 1h 08m 15s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b6bfa779cae4174a7c54c17b76c0502 : SUCCESS in 2h 02m 50s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0e0fd8e978954c029394d95f9e4f2950 : SUCCESS in 2h 53m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6088caff8b1448aff35ef1cf61e7d61363bf5217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-17 19:28:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"de5075d2e2794727a8f68343b7a2c8d956dd1cf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 22:34:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b17b710272974906950923801202a11d : FAILURE in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df104f01ac8a4f11b99b7c3ba4bde665 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b034d5ff23f44415a9d0c2a6b367c3b1 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bb7793a87a243518ee0ed2093f3cd91 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/caa151e9fb2c48fc8a0ae1826536c653 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3684b602d52944cc9da49023324338df : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10df63cdf3f49439756fe8d7ceae8ac : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d84c43c96cc547cc98787e19005bc9a9 : SUCCESS in 7m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bf37ad8fa85245938a8571cee63d29b8 : SUCCESS in 6m 17s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a107ddd360ba4d17824a66495e1f7722 : SUCCESS in 2h 16m 56s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfcf4766e9424ad79c2aa8d92d350c6b : SUCCESS in 1h 45m 53s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/7a85b6a682804a6ca4e89d2a0147306f : TIMED_OUT in 3h 02m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4999fa2cdaa88f78769c73d83dcbfb51d6c52752","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-18 01:13:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"91956f6fede63593944ca567d3b949a2e182eeca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 03:18:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/23a5ce0ca52b454f863cfcd7cd06c1ac : SUCCESS in 7m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f22f7c883ddf4ad191fc63bbe82fb2e0 : FAILURE in 7m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/160fe1099d434376aa3018833c75c877 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31494cd3b3a24ac49b7b157bfe5605f4 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d724be91377a438b8ede3d23f632cbd6 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/652826bc5fd0439ba324b6a8601ac6b3 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8965407df9f42b18a3c97b8b00041d1 : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44d83c417af2439b85bf9592dee3315c : SUCCESS in 5m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c94a93b3487f47a19584787d4bfc17fc : SUCCESS in 6m 39s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/75b83b7327d6450f80161237eac06c03 : SUCCESS in 1h 15m 18s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23243345aadf408a9f8e973f8a0c4fdf : TIMED_OUT in 2h 02m 48s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/fc299001e52447c58bc022eae4c65011 : SUCCESS in 1h 54m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"76e70ec16680841ec9bbef6b62ce73cb01b2aff8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-18 04:16:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1f69c239061a5d19857842fa6d5aac4d8b4ab2a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-18 17:34:43.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f66cd6f2eb2e4cf2e2bfcbadd0109c1f53146f7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 20:49:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/83beee616a2c46d6975e54ce8b7f86d1 : SUCCESS in 8m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5f7a4ca064c475ebfff297a00a833f7 : FAILURE in 7m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ffa31e0b3984a99a6126fb0884a3983 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9393614fc974fd0b73ee1cb6191f12e : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa7b4d6536354da9887693f7d3b1f24f : SUCCESS in 8m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6919ee06a59d4ffaaefcef7e6b5c0868 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d6885ad07241e2aed91045ba14d339 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3eb77c72a5e4d4b8950aa3c7c7c591c : SUCCESS in 7m 23s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/4579c1ff7a5240a4ac75428321dde259 : SUCCESS in 7m 05s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/71f5936fdd0c477dac44e041879efe97 : SUCCESS in 1h 33m 49s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94a8d4361c1b4c4691c754e8b8190468 : SUCCESS in 1h 51m 33s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/38f362e69ee14d10aa34d975c5240f19 : TIMED_OUT in 3h 02m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"66440c57bcc87f71ec7610451986386bc4ad0252","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-02-18 21:39:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e9686c0f673f8ebfeb99cce2de639dc99a31ca6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 21:39:44.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b001d9808d490bc4337709c3b67c7cdbb4c81057","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:22:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/961e4c39217b4ab0af4a8890c9890407 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/683fd077a4604d6fb3153689ef7c04c6 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03b8525abaf24909ba303ba612e9c06f : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4cc727ba9714b5382b8640e546ea570 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38e5cb7f8094594a4d127c3b831e065 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5bacd3122da4a77bbc75ccacefc68e6 : SUCCESS in 6m 10s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f4440e0badcb434d9bdb52095b71a0c1 : SUCCESS in 1h 17m 21s","accounts_in_message":[],"_revision_number":5},{"id":"158fd7ddcdb71427034b44c993ed2e4047559b12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:22:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"9aedcf99648fa9ae3fa24799ca34129397584469","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 23:26:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/1351f577823e4514860d5a0a68c0af08 : SUCCESS in 1m 04s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d14dc6ffa14e9c9e9d2ea8a12d50ae : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f22f38ed5cf248068bc14006d21ef550 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6fdf11ea7f5c77e83dd746fa33b7a354417aec08","revisions":{"e8cb048b7c243ab02b00a54b3e328baec562390f":{"kind":"REWORK","_number":1,"created":"2021-02-16 20:20:48.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/28/775928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/775928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/775928/1"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-15 23:23:19.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-16 20:19:46.000000000","tz":780},"subject":"Support to restore backup from remote location","message":"Support to restore backup from remote location\n\nIn multi-region deployment with geo-replicated Swift, the user can\nrestore a backup in one region by manually specifying the original\nbackup data location created in another region.\n\nChange-Id: Iefef3bf969163af707935445bc23299400dc88c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8cb048b7c243ab02b00a54b3e328baec562390f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e8cb048b7c243ab02b00a54b3e328baec562390f"}]},"branch":"refs/heads/master"},"d5c4fec92ddfdc134c1abab31d2637a23e8cfabc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-17 09:26:23.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/28/775928/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/775928/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/775928/2"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-15 23:23:19.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-17 09:26:18.000000000","tz":0},"subject":"Support to restore backup from remote location","message":"Support to restore backup from remote location\n\nIn multi-region deployment with geo-replicated Swift, the user can\nrestore a backup in one region by manually specifying the original\nbackup data location created in another region.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/775940\nChange-Id: Iefef3bf969163af707935445bc23299400dc88c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5c4fec92ddfdc134c1abab31d2637a23e8cfabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5c4fec92ddfdc134c1abab31d2637a23e8cfabc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d32a25fa1a09bfbba727e93be7583b21daca72da":{"kind":"REWORK","_number":3,"created":"2021-02-17 19:28:21.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/28/775928/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/775928/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/775928/3"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-15 23:23:19.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-17 19:27:42.000000000","tz":780},"subject":"Support to restore backup from remote location","message":"Support to restore backup from remote location\n\nIn multi-region deployment with geo-replicated Swift, the user can\nrestore a backup in one region by manually specifying the original\nbackup data location created in another region.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/775940\nChange-Id: Iefef3bf969163af707935445bc23299400dc88c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d32a25fa1a09bfbba727e93be7583b21daca72da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d32a25fa1a09bfbba727e93be7583b21daca72da"}]},"branch":"refs/heads/master"},"83cbf9b19ec3598eb0b68de14168e19ab00be7a0":{"kind":"REWORK","_number":4,"created":"2021-02-18 01:13:10.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/28/775928/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/775928/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/775928/4"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-15 23:23:19.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-18 01:12:56.000000000","tz":780},"subject":"Support to restore backup from remote location","message":"Support to restore backup from remote location\n\nIn multi-region deployment with geo-replicated Swift, the user can\nrestore a backup in one region by manually specifying the original\nbackup data location created in another region.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/775940\nChange-Id: Iefef3bf969163af707935445bc23299400dc88c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83cbf9b19ec3598eb0b68de14168e19ab00be7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/83cbf9b19ec3598eb0b68de14168e19ab00be7a0"}]},"branch":"refs/heads/master"},"6fdf11ea7f5c77e83dd746fa33b7a354417aec08":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-02-18 17:34:43.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/28/775928/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/775928/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/775928/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/775928/5"}}},"commit":{"parents":[{"commit":"9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5","subject":"Do not override RESTART_REQUIRED service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/9c2e0bf3a0f1bc0b35a148174d8a4d2083f2b3c5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-15 23:23:19.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-02-18 17:34:37.000000000","tz":0},"subject":"Support to restore backup from remote location","message":"Support to restore backup from remote location\n\nIn multi-region deployment with geo-replicated Swift, the user can\nrestore a backup in one region by manually specifying the original\nbackup data location created in another region.\n\nChange-Id: Iefef3bf969163af707935445bc23299400dc88c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fdf11ea7f5c77e83dd746fa33b7a354417aec08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6fdf11ea7f5c77e83dd746fa33b7a354417aec08"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
