)]}'
{"id":"openstack%2Ftrove~781193","triplet_id":"openstack%2Ftrove~master~I0a4a27090ded84aa54c853d78a759f47fcf47ab8","project":"openstack/trove","branch":"master","topic":"remove_context_args","hashtags":[],"change_id":"I0a4a27090ded84aa54c853d78a759f47fcf47ab8","subject":"remove args context never used in clients_admin","status":"ABANDONED","created":"2021-03-18 05:05:35.000000000","updated":"2021-03-18 08:02:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9673df78afe2cab305dd3dec1d12da0abdf4d28c","_number":781193,"virtual_id_number":781193,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-18 06:00:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":-2,"date":"2021-03-18 06:14:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 06:00:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 06:14:28.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"c165caaecf782de588de71522e3f1a658dfb1a85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-03-18 05:05:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95fd3229834c62e863e983d0127535d9893ceee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 06:00:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9b832391f29427bb8b3d0e447093c6c : FAILURE in 6m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc882b46a6a047339f3e555b6728312e : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd2ec25cabb04ef4aeaf98c69867c63d : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50afa5b35b3e4e7fb0a759b3a8eb4c78 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c35752b898d4edca99db49041f23977 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f0e658db64427196f55cca3b3eeda1 : SUCCESS in 6m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a6cbc39546ca40fd979781fdbf5907e7 : SUCCESS in 5m 41s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0457d8bf547d4260a2eb9714afe0c231 : FAILURE in 49m 23s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e16012671377c080c15ad7157b1c3dd5cf44fb9d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-18 06:14:28.000000000","message":"Patch Set 1: Code-Review-2\n\n`context` is a required param (for historical reasons) of the method `create_nova_client`.","accounts_in_message":[],"_revision_number":1},{"id":"870ea0ecc1d33cfe5e1e826028ed73f72670c607","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-03-18 07:18:37.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-2\n\u003e \n\u003e `context` is a required param (for historical reasons) of the method `create_nova_client`.\n\nCan you give more explain for me? I don\u0027t understand why is historical reasons.","accounts_in_message":[],"_revision_number":1},{"id":"15f452fd4f8f2b9f45b69ef2f343c8671a9db75a","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-03-18 07:49:14.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-2\n\u003e \u003e \n\u003e \u003e `context` is a required param (for historical reasons) of the method `create_nova_client`.\n\u003e \n\u003e Can you give more explain for me? I don\u0027t understand why is historical reasons.\n\nyeah, I saw an problem","accounts_in_message":[],"_revision_number":1},{"id":"9673df78afe2cab305dd3dec1d12da0abdf4d28c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-03-18 08:02:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e15820cd13a3dc17213159c1bba8eaffb4da44b","revisions":{"6e15820cd13a3dc17213159c1bba8eaffb4da44b":{"kind":"REWORK","_number":1,"created":"2021-03-18 05:05:35.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/93/781193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/93/781193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/781193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/781193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/93/781193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/93/781193/1"}}},"commit":{"parents":[{"commit":"7dc95b61bc5b7a7f64b7f3206a6f10d4db59ca18","subject":"doc: Restore backup from other regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7dc95b61bc5b7a7f64b7f3206a6f10d4db59ca18"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-03-18 05:05:05.000000000","tz":420},"committer":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2021-03-18 05:05:05.000000000","tz":420},"subject":"remove args context never used in clients_admin","message":"remove args context never used in clients_admin\n\nChange-Id: I0a4a27090ded84aa54c853d78a759f47fcf47ab8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e15820cd13a3dc17213159c1bba8eaffb4da44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6e15820cd13a3dc17213159c1bba8eaffb4da44b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
