)]}'
{"id":"openstack%2Ftrove~782177","triplet_id":"openstack%2Ftrove~master~Ic4bed95dafa6e185a9167c0570695c2b06eb3970","project":"openstack/trove","branch":"master","topic":"add-xena-python-jobtemplates","hashtags":[],"change_id":"Ic4bed95dafa6e185a9167c0570695c2b06eb3970","subject":"Add Python3 xena unit tests","status":"MERGED","created":"2021-03-22 10:27:10.000000000","updated":"2021-03-22 22:57:46.000000000","submitted":"2021-03-22 22:53:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782177-add-xena-python-jobtemplates","meta_rev_id":"3b1fa62abf1b821d0f1293457e82323f65c93945","_number":782177,"virtual_id_number":782177,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 22:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-03-22 21:23:06.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-03-22 21:23:06.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 13:49:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 21:23:06.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"df139d495c976c44d56eff561f99161dd16abf30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-22 10:27:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91377c8c179bdc166c0a5ac16e108cde4aa89a99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 13:49:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/654b08d562f1469f8c3b3a7f3cf335a3 : FAILURE in 8m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/222c104b0362417581b7e0f7cfacc0cc : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8538a6d5f0c8449b8cda438033a8df85 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a52ac8a356e4b68ab6aa69247410f64 : SUCCESS in 9m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b330de4155542e2b1bf109ec3be8a65 : SUCCESS in 10m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3589764ec588461497f83fe0dcd18a3b : SUCCESS in 8m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/55f1d03bf7624507bc7f90ba50e8c5b8 : SUCCESS in 7m 50s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2de1b1f340a8463c976993ebe7075ae4 : SUCCESS in 1h 57m 45s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/870c5e934ac041fbb90eccd573bb37bb : SUCCESS in 1h 50m 24s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/85e33d617eca4cefa2972fa27cb8251e : TIMED_OUT in 3h 03m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"078e947e8005f3518b17b3dfb289fe9edfc02e0b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-22 21:23:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b485c18933aea3a6f4a27ed355e35de80cc9cd57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 21:23:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2e0fe757fe0a49481dfdcd426c20ec3e4c2d7d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 22:53:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd47276ec3c4f29be1eab157d670166 : SUCCESS in 10m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a641eb72f0e4e468cbb05a5aa963e1f : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a8bdd89fbd64b9faf7d1b084d09a389 : SUCCESS in 9m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a893299cce4df6acc4dec298f4ef02 : SUCCESS in 7m 15s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/dc16a311e3814222a617595cc0f0c85b : SUCCESS in 1h 27m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6425ed32b894e04999d81b248ab1727ac8602979","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 22:53:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3b1fa62abf1b821d0f1293457e82323f65c93945","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 22:57:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87c9955ecbb94310b8f08a20c7cb9471 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b057b6ecdd7f4e6a0c659372884502926ab616e2","revisions":{"b057b6ecdd7f4e6a0c659372884502926ab616e2":{"kind":"REWORK","_number":1,"created":"2021-03-22 10:27:10.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/77/782177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/77/782177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/782177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/782177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/77/782177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/77/782177/1"}}},"commit":{"parents":[{"commit":"57ac4a8417ee5c355c36884084b16e61b0c63ba0","subject":"Update master for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/57ac4a8417ee5c355c36884084b16e61b0c63ba0"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-22 10:27:09.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-22 10:27:09.000000000","tz":0},"subject":"Add Python3 xena unit tests","message":"Add Python3 xena unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for xena.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ic4bed95dafa6e185a9167c0570695c2b06eb3970\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b057b6ecdd7f4e6a0c659372884502926ab616e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b057b6ecdd7f4e6a0c659372884502926ab616e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
