)]}'
{"id":"openstack%2Ftrove~783537","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~I7d8a23a2b015d3eead2eefb5df158e950fff0943","project":"openstack/trove","branch":"stable/victoria","topic":"fix-function-test-stable/victoria","hashtags":[],"change_id":"I7d8a23a2b015d3eead2eefb5df158e950fff0943","subject":"Fix test_create_inactive_datastore_by_admin","status":"MERGED","created":"2021-03-27 09:32:32.000000000","updated":"2021-03-28 11:01:48.000000000","submitted":"2021-03-28 11:00:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783537-fix-function-test-stable/victoria","cherry_pick_of_change":782851,"cherry_pick_of_patch_set":1,"meta_rev_id":"f3193953ac5787388075a1331150f6144247bdf5","_number":783537,"virtual_id_number":783537,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-28 11:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-03-28 08:56:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-03-28 08:56:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-27 11:52:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bf176751f4ff167a9e2c4597f6cc2595d9c27e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-27 09:32:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a1723d4f8fcd8eae3b43410108cefd7ba1bfcc11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 11:52:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/802bb1966c674bdc8508643c3b38260d : FAILURE in 7m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1d933c7e8441dd9e471a44fdd2ee75 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/765c1601fda9400087193ee19d27b5b4 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e015aaf7eec4ea6a5449ad40004809e : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb3b0fea6a2437c8f0abaad3c8f90f6 : SUCCESS in 6m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/66965e82652446108cc62cdbd9fe2abb : SUCCESS in 6m 52s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5d57da3230544c5d8fcd87a8eeb99191 : TIMED_OUT in 2h 12m 55s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82b510242c81445db67aa629fd0a8f74 : TIMED_OUT in 2h 02m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a502e22b507c8aa4a5b54e875b606ff71a02348e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-28 08:56:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e614dc545eef75e1ef221f4445394c15c950766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-28 08:56:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89c36bf42732ed6d036397c31d9dbdb3bcd7b449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-28 11:00:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8a85bd7663445b1bf69bb44e0c32303 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/152a5237b1654658b06cc62f6fc326f8 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39bf7690994e4407baa1fd49731d7dee : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ac23888826437e9d56746427005c49 : SUCCESS in 7m 00s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/0810e71810b74cf39aae3bb15ce8805c : FAILURE in 2h 03m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08cabf5bc2f7998571e9732f30554a68c8b5903c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-28 11:00:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f3193953ac5787388075a1331150f6144247bdf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-28 11:01:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acaf5f95694045cf87c6ae2c73ceda1b : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"368915a29d8bfe89861ffd3c1b7bb1f3b899416f","revisions":{"368915a29d8bfe89861ffd3c1b7bb1f3b899416f":{"kind":"REWORK","_number":1,"created":"2021-03-27 09:32:32.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/37/783537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/783537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/783537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/783537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/783537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/783537/1"}}},"commit":{"parents":[{"commit":"ed5c5f962dfdb03593411c52113cee12979294f7","subject":"Stop using pip-and-virtualenv diskimage element","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ed5c5f962dfdb03593411c52113cee12979294f7"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-24 19:49:05.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-27 09:32:32.000000000","tz":0},"subject":"Fix test_create_inactive_datastore_by_admin","message":"Fix test_create_inactive_datastore_by_admin\n\nChange-Id: I7d8a23a2b015d3eead2eefb5df158e950fff0943\n(cherry picked from commit e4b70cf73a7e8069f36273690e630022dbabab67)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/368915a29d8bfe89861ffd3c1b7bb1f3b899416f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/368915a29d8bfe89861ffd3c1b7bb1f3b899416f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
