)]}'
{"id":"openstack%2Ftrove~783762","triplet_id":"openstack%2Ftrove~master~Iafe870e32a1b870e4025e08760c8c3ebeb67dab5","project":"openstack/trove","branch":"master","topic":"add-job-trove-tempest-postgresx","hashtags":[],"change_id":"Iafe870e32a1b870e4025e08760c8c3ebeb67dab5","subject":"Add CI job trove-tempest-postgres","status":"MERGED","created":"2021-03-30 00:46:19.000000000","updated":"2021-03-30 21:51:33.000000000","submitted":"2021-03-30 21:43:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783762-add-job-trove-tempest-postgresx","meta_rev_id":"3a0c5eb9766d3133f0f05393b8da6600387b6d5b","_number":783762,"virtual_id_number":783762,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-30 21:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-30 20:17:09.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-30 20:17:09.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 00:46:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5c95bbe63caf6df4f5bfc228b7385c726ae9fcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-30 00:46:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a4a13ed4e129841bac1619e330637a3e0d590eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 00:46:36.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/trove/.zuul.yaml@master\", line 306, column 24\nsecond occurrence\n  in \"openstack/trove/.zuul.yaml@master\", line 358, column 24","accounts_in_message":[],"_revision_number":1},{"id":"e12691ce359bc3c153d303a5821c0a1d8dd7f13a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-30 00:48:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f74f5300008dbfd3530ac2a882b03da1be7a06f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 00:49:14.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/trove/.zuul.yaml@master\", line 306, column 24\nsecond occurrence\n  in \"openstack/trove/.zuul.yaml@master\", line 358, column 24","accounts_in_message":[],"_revision_number":2},{"id":"27fbb03c5c92ee23c700ebfc16ff753e6ef947fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-30 01:06:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7be06b215074e19271baf97822549f99d58978c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 01:06:28.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch master.  The error was:\n\n  found duplicate anchor; first occurrence\n  in \"openstack/trove/.zuul.yaml@master\", line 316, column 11\nsecond occurrence\n  in \"openstack/trove/.zuul.yaml@master\", line 370, column 11","accounts_in_message":[],"_revision_number":3},{"id":"881c19ccadc15af2ef768b9408d7a701eaba0353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-30 01:31:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b902ceabcc5153fc546335dedb3af876628a0e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 03:50:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43bb2c30c589420295ca66f3228361c3 : FAILURE in 7m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8d9ea2cdbfa4ef7b17aa53a456e8bbc : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44f6fa460f9640d89e616efd5dcca02f : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9353632897ce4adbbbb885b1d6d72ff3 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6153399b9da941a79a298f447427abf6 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1c75657de114a84813b6f7d0a5ca928 : SUCCESS in 6m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7dd452bde813432eaf6bf588dfd8dfce : SUCCESS in 6m 28s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/7972521e00794747b116c88243329ad0 : FAILURE in 1h 39m 24s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/b8cf7cf33d7a4e6c8fbb21e66bd33af8 : SUCCESS in 1h 20m 57s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8afea4139ac5461485c81502bf5b11fc : TIMED_OUT in 2h 03m 08s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/276561d1bfe74a45b8c2766f849a1a84 : SUCCESS in 1h 51m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0326acb57df6f8541f381a00fc3ca227318b4883","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-30 05:28:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e118be8396513836da5d3a4363fb442331c39fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 08:38:09.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6da52fb1a48e44449147055394524dfe : FAILURE in 7m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc364cfc644f4e0cafeef34c2ecef27b : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73b6153d06f443f097bbd33a2df1953c : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b179e639afc14bd4bf56f919d2136534 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e14e141d6de478ba6bc63c77af77a1f : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b56544fc9746d0ac6b128524b69fa9 : SUCCESS in 5m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/817e96b5432840cead403c2c1e1637c5 : SUCCESS in 6m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2b5b1e106e754cd2bfaa7da549e147cf : SUCCESS in 1h 15m 06s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d34a7f290adf46c0ac6912993a3aea13 : FAILURE in 1h 44m 15s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65cefc582ab446bdac353cbe69289c5d : SUCCESS in 1h 58m 19s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4d7ddeed79ef4ddeadef01091aec2304 : TIMED_OUT in 3h 03m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b82c736f0723be3eb776b918d51f7a720c4dca7a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-30 09:18:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1c596e192fccef4f78befff3e898cb9a76ec8913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 12:08:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04fd6fd3bb7d44edb6494e362be7debf : FAILURE in 9m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919166ca5c334fed92ab2bf6c908f8eb : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b37e5ae7f3404ab0904f067cb963f1e9 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0425e50e972b4bf0b956517fcdd6f5cc : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64f23a3afeb04135a057463ed0a19924 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ee79ea94160402ebcecb25a1d9914bf : SUCCESS in 6m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/472f4db02a774b52a6a309cf803ed330 : SUCCESS in 7m 10s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9f7f50ed417f4650ba275135663309a9 : SUCCESS in 1h 11m 27s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/707d4686781c4747bd5dcb71a0891a57 : SUCCESS in 1h 57m 05s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abe62f24eb5745278ca33064f52280d8 : SUCCESS in 2h 01m 42s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e85ccfff0e804b53a5117826de4ce811 : SUCCESS in 2h 47m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cb064dd13e5e02390bc80c37d7d8b42f92c062fe","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-03-30 20:17:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5dcfbe13c078394fbaa62689eeecc78944323373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:17:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"308045822ffc13a12cfd771bdc66c86168fb7944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 21:43:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ea820ef69b741b99678218e4a0531f7 : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22b03a1d98324ce69cb01d2910cbd5c1 : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce2b651fb38d4c70bf8ff3d38e3d0d17 : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae95f84750594b1096da9fad2f894d54 : SUCCESS in 9m 21s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a999ddedbba04000a69f16b6679649ba : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":4},{"id":"f2200b5fadcafec85c571b350eeb2d8b724b44d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 21:43:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3a0c5eb9766d3133f0f05393b8da6600387b6d5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 21:51:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d847d073c02e4103a350f5efa6c1a591 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1f1a1e4ee4bc880131b76241d61903f3a72d0b40","revisions":{"abfbf415c8324813e8c642b64f5afb1286421958":{"kind":"REWORK","_number":1,"created":"2021-03-30 00:46:19.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/62/783762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/62/783762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/62/783762/1"}}},"commit":{"parents":[{"commit":"3bf2e73f47a31661c16888ecae79640528432211","subject":"Merge \"Improve guest agent when start\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3bf2e73f47a31661c16888ecae79640528432211"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-30 00:46:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-30 00:46:09.000000000","tz":780},"subject":"Add CI job trove-tempest-postgres","message":"Add CI job trove-tempest-postgres\n\nChange-Id: Iafe870e32a1b870e4025e08760c8c3ebeb67dab5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/abfbf415c8324813e8c642b64f5afb1286421958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/abfbf415c8324813e8c642b64f5afb1286421958"}]},"branch":"refs/heads/master"},"7d7e819a5244c7b5b04976b87d8d7a1ac29c96f0":{"kind":"REWORK","_number":2,"created":"2021-03-30 00:48:48.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/62/783762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/62/783762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/62/783762/2"}}},"commit":{"parents":[{"commit":"3bf2e73f47a31661c16888ecae79640528432211","subject":"Merge \"Improve guest agent when start\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3bf2e73f47a31661c16888ecae79640528432211"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-30 00:46:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-30 00:48:33.000000000","tz":780},"subject":"Add CI job trove-tempest-postgres","message":"Add CI job trove-tempest-postgres\n\nChange-Id: Iafe870e32a1b870e4025e08760c8c3ebeb67dab5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d7e819a5244c7b5b04976b87d8d7a1ac29c96f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7d7e819a5244c7b5b04976b87d8d7a1ac29c96f0"}]},"branch":"refs/heads/master"},"2a39ec1e3f103edb4538f9872e15909526ca4bf6":{"kind":"REWORK","_number":3,"created":"2021-03-30 01:06:05.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/62/783762/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/62/783762/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/62/783762/3"}}},"commit":{"parents":[{"commit":"3bf2e73f47a31661c16888ecae79640528432211","subject":"Merge \"Improve guest agent when start\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3bf2e73f47a31661c16888ecae79640528432211"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-30 00:46:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-30 01:05:56.000000000","tz":780},"subject":"Add CI job trove-tempest-postgres","message":"Add CI job trove-tempest-postgres\n\nChange-Id: Iafe870e32a1b870e4025e08760c8c3ebeb67dab5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2a39ec1e3f103edb4538f9872e15909526ca4bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2a39ec1e3f103edb4538f9872e15909526ca4bf6"}]},"branch":"refs/heads/master"},"1f1a1e4ee4bc880131b76241d61903f3a72d0b40":{"kind":"REWORK","_number":4,"created":"2021-03-30 01:31:04.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/62/783762/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/62/783762/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/62/783762/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/62/783762/4"}}},"commit":{"parents":[{"commit":"3bf2e73f47a31661c16888ecae79640528432211","subject":"Merge \"Improve guest agent when start\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3bf2e73f47a31661c16888ecae79640528432211"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-30 00:46:09.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-03-30 01:30:54.000000000","tz":780},"subject":"Add CI job trove-tempest-postgres","message":"Add CI job trove-tempest-postgres\n\nChange-Id: Iafe870e32a1b870e4025e08760c8c3ebeb67dab5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f1a1e4ee4bc880131b76241d61903f3a72d0b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f1a1e4ee4bc880131b76241d61903f3a72d0b40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
