)]}'
{"id":"openstack%2Ftrove~784063","triplet_id":"openstack%2Ftrove~master~I0927466d18eca5572932fcf9e2ca6349b11ca847","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I0927466d18eca5572932fcf9e2ca6349b11ca847","subject":"[doc] Add configuration reference","status":"MERGED","created":"2021-03-31 10:11:20.000000000","updated":"2021-04-06 23:58:58.000000000","submitted":"2021-04-06 23:57:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"784063","meta_rev_id":"22ae8166094a10cd0f1825d37ea651c104accfb2","_number":784063,"virtual_id_number":784063,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 23:57:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-06 22:38:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-06 22:38:35.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 12:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-31 17:46:58.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"975530ade3d3a0ef02c4ec7e7ff1b6b25024a971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 10:11:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"822012039247a0a17ae3ad3d3aa9fd98ff167b30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 12:22:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/342efa97a9b64ac08024f582ffcdd238 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8624d2bcd50c4cc5b1b58eda99ac6c83 : FAILURE in 7m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da82b099f7754d03a26946e026372f38 : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddd68c1e5fe74c5dabc1f3f5148de546 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/260e7eacf7664b5eaa2716ef37ab66dd : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8926036351b94470a170b5086b7d98d6 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/223d9c72bdc645b5a59adb6f769555b3 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc9efd70638245e192df039f39f7fdd0 : SUCCESS in 7m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/5212c1de5c6a412da191b28f103680bb : SUCCESS in 8m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4350004f36a94533a7c03fa91129e27b : SUCCESS in 1h 54m 20s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/8fdb4f4e4c374338b607e1671eceded8 : SUCCESS in 1h 38m 06s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79c234e670864ec5aeb85e8fa973ee09 : FAILURE in 1h 36m 39s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6d5e0932076847bc8f3ce0a8c82662fa : FAILURE in 1h 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64083cb4cddd5a63ece1fd2f331f8727d99402c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-31 13:26:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f09516b6c46082b4aaf92d8dccabab06cc2321f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:38:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/454076dcc67441baa1481f4a40ec2983 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/657dcd492c5944ac98091e3e21c70b43 : FAILURE in 7m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c097da5553344577860c8356afc14357 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef76baadd14147ed90dbf2bb71293095 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fda29a73bdf473e867c6d2d4fca69cb : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74a744f26c8b49d9a8ca70d902fc5ba3 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2e41b646ca495cb5200c96a47ea9e3 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d07a2aebf4946408310b0a054ef5db3 : SUCCESS in 6m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/699813695aa34eada598fd6488a06249 : SUCCESS in 6m 14s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ac0ccc6f8bba494bb88a2d31c59fe9c6 : SUCCESS in 1h 28m 21s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/bcc97a1ce0294727920322e19a3ac582 : SUCCESS in 1h 40m 30s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c72b34885a134c41a43cc2bec73cc7ea : SUCCESS in 2h 03m 30s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e1eb5e27ac064b948f04ba0214f0f4cd : SUCCESS in 2h 57m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"813c39f306b651b91873f1be2a6d75aea974282e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-04-06 03:10:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23d6f9356d34e6e7e2b692420617ed30c264f260","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-06 06:35:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"45b5634f86acd5890c166acac45f8dd2c29874d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-06 06:38:20.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"00a77ff40b0220a53a0d772d83296c9e3cd7c8cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 09:07:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/655a6b988560489c8dcb15294f186237 : SUCCESS in 8m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1405306f65d44c8a946a822d9b2131d : FAILURE in 8m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b779a69def4062b6ce52fd4323f50e : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cbbc4e3a0874d139c515fe29cee7678 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbf18851d80c41aeb83b5f4649026972 : SUCCESS in 7m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5d3578adea24df78f96f17c8b754474 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9f108339244cd9a2503529d75bcaba : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/429e0efb7d7c4db8878817c3287c0f07 : SUCCESS in 7m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/daed49687dfc4c5cb488cc7709be5832 : SUCCESS in 7m 39s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/30edd437ec3e433d9d2d913d0463ba19 : SUCCESS in 1h 13m 29s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/22de3c8947c6463c816ad7d16061c2aa : SUCCESS in 1h 03m 37s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/704591f220c24ef0a2f16bd3e16c9618 : TIMED_OUT in 2h 03m 34s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/62a21bc175f64c248ded054bbe5b3521 : SUCCESS in 2h 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c4bb0bf967743b34b34ac955711258fbe8947d99","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-04-06 22:38:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"beb9757bcdd5a47f57494079ccd8ea117e8e44a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 22:38:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"90d398f050c102a22214069c4d17f82f892ac132","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 23:57:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9bc3884a046c4a008bd096e5fa7f3fd1 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/443b2d0cf48a4f269a10cabf87a20abf : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43d9d8049b6c442ba143c6b671154108 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73c1b00b4ad646bc944680224eaefc44 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/742ea446b074444e9e2761d31ec8e941 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48c7df5d75934820add43a41515130a4 : SUCCESS in 5m 31s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/8531833b8073420e80060e30ad7eef9b : SUCCESS in 1h 14m 02s","accounts_in_message":[],"_revision_number":3},{"id":"c8ed15b484a92ccd126f43d5629d534fa4198e67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 23:57:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"22ae8166094a10cd0f1825d37ea651c104accfb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 23:58:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c2faf297c50244d79db4ecf18c65b3fb : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4991736ed864aabbbb8ff20331c36b7 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9578e9235d7d4a59a1194796dc0f61b1 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dfca9b1943d1c64533ef754bd3c248dacdd3d627","revisions":{"3d15d30a1ddab7407e778bdf9cd4edfdd7ad4f97":{"kind":"REWORK","_number":1,"created":"2021-03-31 10:11:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/784063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/63/784063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/784063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/784063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/784063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/63/784063/1"}}},"commit":{"parents":[{"commit":"1f1a1e4ee4bc880131b76241d61903f3a72d0b40","subject":"Add CI job trove-tempest-postgres","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f1a1e4ee4bc880131b76241d61903f3a72d0b40"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-31 10:10:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-31 10:11:16.000000000","tz":180},"subject":"[doc] Add configuration reference","message":"[doc] Add configuration reference\n\nConfig reference has been dropped for some reason several releases before.\nWe return configuration reference along with policy to the docs.\n\nChange-Id: I0927466d18eca5572932fcf9e2ca6349b11ca847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3d15d30a1ddab7407e778bdf9cd4edfdd7ad4f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3d15d30a1ddab7407e778bdf9cd4edfdd7ad4f97"}]},"branch":"refs/heads/master"},"a4cebe1e14e3632613a09bc76495c298e4c78a15":{"kind":"REWORK","_number":2,"created":"2021-03-31 13:26:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/784063/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/63/784063/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/784063/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/784063/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/784063/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/63/784063/2"}}},"commit":{"parents":[{"commit":"1f1a1e4ee4bc880131b76241d61903f3a72d0b40","subject":"Add CI job trove-tempest-postgres","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f1a1e4ee4bc880131b76241d61903f3a72d0b40"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-31 10:10:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-31 13:25:52.000000000","tz":180},"subject":"[doc] Add configuration reference","message":"[doc] Add configuration reference\n\nConfig reference has been dropped for some reason several releases before.\nWe return configuration reference along with policy to the docs.\n\nChange-Id: I0927466d18eca5572932fcf9e2ca6349b11ca847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4cebe1e14e3632613a09bc76495c298e4c78a15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a4cebe1e14e3632613a09bc76495c298e4c78a15"}]},"branch":"refs/heads/master"},"dfca9b1943d1c64533ef754bd3c248dacdd3d627":{"kind":"REWORK","_number":3,"created":"2021-04-06 06:38:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/63/784063/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/63/784063/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/784063/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/784063/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/63/784063/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/63/784063/3"}}},"commit":{"parents":[{"commit":"1f1a1e4ee4bc880131b76241d61903f3a72d0b40","subject":"Add CI job trove-tempest-postgres","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1f1a1e4ee4bc880131b76241d61903f3a72d0b40"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-31 10:10:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-06 06:37:59.000000000","tz":0},"subject":"[doc] Add configuration reference","message":"[doc] Add configuration reference\n\nConfig reference has been dropped for some reason several releases before.\nWe return configuration reference along with policy to the docs.\n\nChange-Id: I0927466d18eca5572932fcf9e2ca6349b11ca847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dfca9b1943d1c64533ef754bd3c248dacdd3d627"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/dfca9b1943d1c64533ef754bd3c248dacdd3d627"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
