)]}'
{"id":"openstack%2Ftrove~786051","triplet_id":"openstack%2Ftrove~master~Ie8b3884dd13147f29d5632275fa5c75e801e04c8","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"Ie8b3884dd13147f29d5632275fa5c75e801e04c8","subject":"Fix check_subnet_router filters","status":"MERGED","created":"2021-04-13 13:31:47.000000000","updated":"2021-04-14 13:22:15.000000000","submitted":"2021-04-14 13:17:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786051","meta_rev_id":"4395863bd85bd0b9ab257006a6dbdfeaeb10da6e","_number":786051,"virtual_id_number":786051,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 13:17:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-04-14 11:37:12.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-04-14 11:37:12.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-13 13:58:24.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2021-04-13 15:41:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86a351237c519841497206761760be1fef32fac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-13 13:31:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28e4f41c2f5217aea6e3607dd821a8b3dc9548ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 15:41:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e02496cc5594074bfbef4f54ef2afab : FAILURE in 8m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e4a79938fb41298b041ecec50d83bd : SUCCESS in 8m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/066c195ce9794298b0969c59f72d54bb : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fec4362119a742678280e351a32365af : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5b29ac0cc764510a688f995bb7f1524 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4b325f6d01406a9417c8707b5e59d0 : SUCCESS in 8m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/694ac1b2f7bf4fef964c51e14eb7e161 : SUCCESS in 6m 35s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/16ce20e29b374386b588afc7aae0133f : FAILURE in 52m 59s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/389bba497aa24c968c4a5d7d7f8a8238 : FAILURE in 1h 34m 59s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/baa14612482848c0aea788dd6e21bc06 : TIMED_OUT in 2h 03m 55s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/dd4bb83ab40c4956b6faddd364c36461 : FAILURE in 31m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38e70c03cf3ec3dae952fae91995e5c4688a55ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-14 07:10:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a11a4c2318d9e5c67d21f422245c86aef0a08b88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 08:30:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a77166f1fca4e7bad8082cf4569bfdf : FAILURE in 7m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b204854dc594eed938efb5c21dd3f12 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08dcd6185c694882854de85ce82cde5c : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c24faf8d5dea4f47aafd0bf42d9154b9 : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b58dd4a8122647299eab2ddfa3b4b500 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c8eb4c96794da29ab9a7bc4ebdf905 : SUCCESS in 6m 26s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d09dc8eefca4aeeb069276e6f1b9bf8 : SUCCESS in 8m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/6d5aa2185c5b4c0e95f86f3072765a3d : SUCCESS in 1h 16m 30s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d9604ac498e444e09ee999c2cbd21922 : SUCCESS in 1h 17m 17s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a333b529268421f98a93e893064840d : FAILURE in 48m 49s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/d43fc609f198484da3c0cce7d38f7047 : FAILURE in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5291bb9b805c97685b70c8e0dcb92971504aabf9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-04-14 11:37:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb4c96ad7d8a41c4bf62d04988d03705f2fde0ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 11:37:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1a8192b40777ff9ff06a490f3e923c63dd5cdfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 13:17:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf4a632842e54fe7b55e9008fc7d7cf8 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc857ab0081541fc8616a617bb7bed74 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a5fcbba6961453d9e5cf6205d261f05 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e1d6fd26b646d5a7d880f4c86fbe58 : SUCCESS in 6m 19s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/605335a9d58f461d96e4060d80acd311 : SUCCESS in 1h 32m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7627000793765b9412a7214fa9e9fafad832c1ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 13:17:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4395863bd85bd0b9ab257006a6dbdfeaeb10da6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 13:22:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c0ee5d92bb4f29b25114efbd029c36 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5497179d4789ef1c22ad8f9a39c719459241a62","revisions":{"d5497179d4789ef1c22ad8f9a39c719459241a62":{"kind":"REWORK","_number":1,"created":"2021-04-13 13:31:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/51/786051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/51/786051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/786051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/786051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/51/786051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/51/786051/1"}}},"commit":{"parents":[{"commit":"22b99fb05d405c893ae75d28a7b4ee1c948a9c63","subject":"Add taskmanager_manager option to common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/22b99fb05d405c893ae75d28a7b4ee1c948a9c63"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-04-13 13:31:33.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-04-13 13:31:33.000000000","tz":180},"subject":"Fix check_subnet_router filters","message":"Fix check_subnet_router filters\n\nDevice owner of router ports is not limited to network:router_interface\nand can also be network:router_interface_distributed or\nnetwork:ha_router_replicated_interface depending on the router\nto which port is attached.\n\nCurrently, when HA router is used, trove fails with:\n`Subnet is not associated with router` since port does not pass filter\n\nChange-Id: Ie8b3884dd13147f29d5632275fa5c75e801e04c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5497179d4789ef1c22ad8f9a39c719459241a62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5497179d4789ef1c22ad8f9a39c719459241a62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
