)]}'
{"id":"openstack%2Ftrove~786222","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~Ie8b3884dd13147f29d5632275fa5c75e801e04c8","project":"openstack/trove","branch":"stable/victoria","hashtags":[],"change_id":"Ie8b3884dd13147f29d5632275fa5c75e801e04c8","subject":"Fix check_subnet_router filters","status":"MERGED","created":"2021-04-14 13:28:11.000000000","updated":"2021-04-14 21:42:59.000000000","submitted":"2021-04-14 21:41:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786222","cherry_pick_of_change":786221,"cherry_pick_of_patch_set":1,"meta_rev_id":"e5a924273eb0257e467993f3c59107d4e31884df","_number":786222,"virtual_id_number":786222,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 21:41:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-04-14 19:26:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-04-14 19:26:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 15:46:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-14 19:26:50.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"12aebf529027533b1acda3a741d0278609c6a90b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-14 13:28:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"87c902baa56c4f8ec18cb738fd7c48c8291eebbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 15:46:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1acaa25c2894499e8079fbe0e9a2f0c6 : FAILURE in 10m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34c719254374748b8e9f7be795d2046 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a60e4bad96354b1e96961c0db8cbba9a : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb5fa20a80fd412687fb9c4701cebd80 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b48043f93af42cca1f72fd742108e28 : SUCCESS in 5m 59s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/193c347340a9462881cf4b69e956d286 : SUCCESS in 6m 24s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1ec938af27054cd699e7ed8116cd9556 : TIMED_OUT in 2h 12m 42s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9bbb75bf7e84198adcb173e847b7dc9 : TIMED_OUT in 2h 02m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42ade2a71c46f7d7f4d83365eece4d6e982a8bab","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-04-14 19:26:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aca6d75969e0dec3cac0507725b8a1a79a73a49d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 19:27:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb75b6f27b37c0e934e56735797749a0d9c7a376","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 21:41:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc671827c81a4abd8cdcda9528277f1c : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16e4aedb71c44d94a1d7adf0a57cb5b0 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/377a28c3df2c4365b11757a87012f008 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ae1c556fcf4304b7225ba13eca46f2 : SUCCESS in 6m 17s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/82d61b5c52ba43688760aa73f993e14d : TIMED_OUT in 2h 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4857d50a85b80f00c662be2185802991d9c73792","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 21:41:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e5a924273eb0257e467993f3c59107d4e31884df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 21:42:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82a5629f683a494993f644abd35ba3e5 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"003b98a73e5da99eadcc1806f3e0cc9a109ea1cb","revisions":{"003b98a73e5da99eadcc1806f3e0cc9a109ea1cb":{"kind":"REWORK","_number":1,"created":"2021-04-14 13:28:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/22/786222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/22/786222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/786222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/786222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/786222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/22/786222/1"}}},"commit":{"parents":[{"commit":"368915a29d8bfe89861ffd3c1b7bb1f3b899416f","subject":"Fix test_create_inactive_datastore_by_admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/368915a29d8bfe89861ffd3c1b7bb1f3b899416f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-04-13 13:31:33.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-04-14 13:28:11.000000000","tz":0},"subject":"Fix check_subnet_router filters","message":"Fix check_subnet_router filters\n\nDevice owner of router ports is not limited to network:router_interface\nand can also be network:router_interface_distributed or\nnetwork:ha_router_replicated_interface depending on the router\nto which port is attached.\n\nCurrently, when HA router is used, trove fails with:\n`Subnet is not associated with router` since port does not pass filter\n\nChange-Id: Ie8b3884dd13147f29d5632275fa5c75e801e04c8\n(cherry picked from commit d5497179d4789ef1c22ad8f9a39c719459241a62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/003b98a73e5da99eadcc1806f3e0cc9a109ea1cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/003b98a73e5da99eadcc1806f3e0cc9a109ea1cb"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
