)]}'
{"id":"openstack%2Ftrove~786765","triplet_id":"openstack%2Ftrove~master~I9055b50d981274abaae248e030d72d9a3facbfef","project":"openstack/trove","branch":"master","topic":"fix-error-create-replicas","hashtags":[],"change_id":"I9055b50d981274abaae248e030d72d9a3facbfef","subject":"Fix error race condition when create replicas for database instances with mariadb datastore manager","status":"ABANDONED","created":"2021-04-18 05:53:35.000000000","updated":"2021-09-15 02:36:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"baacbda6782ab7b2db779b55c054d72d33d743ee","_number":786765,"virtual_id_number":786765,"owner":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-18 08:20:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-18 08:20:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-20 00:50:21.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"CC"}],"messages":[{"id":"7cef662b2985d94d9ac754f5f58e7931da398376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-04-18 05:53:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"184d9db811d69893acc112b22a5bafd53fa035ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 08:20:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6472bba2e8f24756a596a7f3d1fb4418 : FAILURE in 12m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d52b35ced0844f9b6d52bfea7adbc86 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f3d855f94aa4429a73a82f09d9f5eeb : SUCCESS in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6195f9bf638b458880c057299f2829e9 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bf252105ae646b9bbc2aca60e594db5 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131837f1c5334571a546011a15c93f2e : SUCCESS in 5m 19s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9afd242cb3564930bb7c2e8c9343804d : SUCCESS in 6m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/f05edf737e77420087d5e6d2b4e27cb8 : SUCCESS in 1h 14m 54s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/5720bed60e9542db9d7ecf6caedd8af4 : SUCCESS in 1h 19m 33s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caa57223aee1409e8f45d84319b4ca25 : FAILURE in 2h 02m 02s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ee5b6890963b409f909e8dfddb7493df : SUCCESS in 2h 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0f8acb2fa95cd36ed928fa9376eaad543fc69c7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-04-20 00:50:21.000000000","message":"Patch Set 1:\n\nCould you please take a look at my comments in the task?","accounts_in_message":[],"_revision_number":1},{"id":"baacbda6782ab7b2db779b55c054d72d33d743ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2021-09-15 02:36:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"298b99af77ec5dbd06b9e222087241970c9b5927","revisions":{"298b99af77ec5dbd06b9e222087241970c9b5927":{"kind":"REWORK","_number":1,"created":"2021-04-18 05:53:35.000000000","uploader":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"ref":"refs/changes/65/786765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/65/786765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/786765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/786765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/65/786765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/65/786765/1"}}},"commit":{"parents":[{"commit":"d5497179d4789ef1c22ad8f9a39c719459241a62","subject":"Fix check_subnet_router filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5497179d4789ef1c22ad8f9a39c719459241a62"}]}],"author":{"name":"Bo Tran","email":"yeuconay.caconaynua.96.nd@gmail.com","date":"2021-04-18 05:52:15.000000000","tz":420},"committer":{"name":"Bo Tran","email":"yeuconay.caconaynua.96.nd@gmail.com","date":"2021-04-18 05:52:15.000000000","tz":420},"subject":"Fix error race condition when create replicas for database instances with mariadb datastore manager","message":"Fix error race condition when create replicas for database instances\nwith mariadb datastore manager\n\nTask: #42295\nStory: #2008826\nChange-Id: I9055b50d981274abaae248e030d72d9a3facbfef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/298b99af77ec5dbd06b9e222087241970c9b5927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/298b99af77ec5dbd06b9e222087241970c9b5927"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
