)]}'
{"id":"openstack%2Ftrove~789115","triplet_id":"openstack%2Ftrove~master~I7261114a545c8c016f96ecf20e3fcdc00c3bab99","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I7261114a545c8c016f96ecf20e3fcdc00c3bab99","subject":"Dropping lower constraints testing","status":"MERGED","created":"2021-05-03 01:28:49.000000000","updated":"2021-05-03 04:41:05.000000000","submitted":"2021-05-03 04:39:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789115","meta_rev_id":"f4c7716e9121001091ecaf85f1eca09215f1df57","_number":789115,"virtual_id_number":789115,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-03 04:39:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-05-03 01:33:22.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-05-03 01:33:22.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 01:33:22.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2021-05-03 03:34:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce3c0190a6c876cf9dbea6e3ef205d2d0402d8b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-05-03 01:28:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"967fc3de6e70c7ecd9e6a00ef1e9c4a1844667c8","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-05-03 01:33:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62b7d741e118048e9f61362bf776f71ad6adca8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 03:34:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f629806c50f145f9a45a9af4d3fd6cb1 : SUCCESS in 6m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7fb7436570c9465883a2efe356a13681 : SUCCESS in 4m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6ac57ec9488459baa972587d0d2b7a3 : FAILURE in 7m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f9bbbcc5c34b949f382b0186d306f4 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/928644d5feb7411c87d8ab3291776f2b : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acec632efb194196bef1169ea81914b1 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d887267834af4c1bbdeb56a5e2bfdb44 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faff9d553ed54a3691767a9c787db63a : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ad90e53bac747c39154e88e83b2ee02 : SUCCESS in 6m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a45f08e13e5c4c2bb62df1823c0ca824 : SUCCESS in 6m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b4097bb1422e4a38af4f86817555082b : SUCCESS in 1h 12m 55s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d297a6a675ee4dbaacdc5c53459a62a5 : SUCCESS in 1h 16m 34s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbfddd037d954dcc8029c633a4ff6030 : TIMED_OUT in 2h 03m 52s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2636ec3cd09740e9bfd8e0ed8395047d : FAILURE in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42d3633c2223aba651e2f143bed9c70c9595b0dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 03:34:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5add8a757d6cc3b7d49758ec5fd533f88e7fa74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 04:39:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/91da8f475dbf47b1b71bbde33215c046 : SUCCESS in 6m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a281b79b76ad4d56b8a8d472c1f83db0 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e9da3de5a041dfa5784088a1faebe8 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4786e2911b414f79bf75816995498806 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2cc1e4d2e3d43ad94bb5aa070c8ae46 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/265f93120e1a403789b676d5786ee86c : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/369f78a6d87448558b92a686e82acdac : SUCCESS in 6m 28s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/18aff1a0f69d4c0ea58c35084f315c91 : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":1},{"id":"80e59d2d36ba2077360c86371e4ee7bdf80ae590","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 04:39:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f4c7716e9121001091ecaf85f1eca09215f1df57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 04:41:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/4e9bf9f04f9946869ab6330794557872 : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1da9090d0e146b69edaa39d1e035f57 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aed5d770e61f45dea52c44efb823fc95 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"070f82f8cab3434cf81cf584193f0cbd289476ab","revisions":{"070f82f8cab3434cf81cf584193f0cbd289476ab":{"kind":"REWORK","_number":1,"created":"2021-05-03 01:28:49.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/15/789115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/789115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/789115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/789115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/789115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/789115/1"}}},"commit":{"parents":[{"commit":"d5497179d4789ef1c22ad8f9a39c719459241a62","subject":"Fix check_subnet_router filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5497179d4789ef1c22ad8f9a39c719459241a62"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-05-03 01:23:03.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2021-05-03 01:28:44.000000000","tz":480},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain needed\nto maintain that now exceeds the benefits provided by this mechanismes.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: I7261114a545c8c016f96ecf20e3fcdc00c3bab99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/070f82f8cab3434cf81cf584193f0cbd289476ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/070f82f8cab3434cf81cf584193f0cbd289476ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
