)]}'
{"id":"openstack%2Ftrove~789116","triplet_id":"openstack%2Ftrove~master~I41d989d876a73f49684d8212d27e48876190f210","project":"openstack/trove","branch":"master","hashtags":[],"change_id":"I41d989d876a73f49684d8212d27e48876190f210","subject":"setup.cfg: add oslo.policy.enforcer","status":"MERGED","created":"2021-05-03 01:32:58.000000000","updated":"2021-05-04 03:54:13.000000000","submitted":"2021-05-04 03:53:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789116","meta_rev_id":"4a806037885c73908e70daa4d381aa609873f63d","_number":789116,"virtual_id_number":789116,"owner":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-04 03:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-05-04 02:27:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-05-04 02:27:37.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 04:37:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-04 02:27:37.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"f77f66b2666d25c559375b8b7e9813b9a4665dc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2021-05-03 01:32:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"266e203571de545070f5fae74415562f3b9d47aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 04:37:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/214c4a6e49d347d1b463738d60191921 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddd05cf2b0f94cc998e09fd59ee9b963 : FAILURE in 10m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69277ec357b34d799c3e596d41495c1e : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e828c5269e249ae888d769f4db3952f : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/249d1503eafc4bb6b699a0d0f4309475 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfecd0908f4b41b99ec8d08ed6d256fd : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1adbf47fd39e4ed9a0c61f063d36d15c : SUCCESS in 6m 10s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5f0c5d62f75b4888a3d953c4c55ef45d : SUCCESS in 1h 03m 16s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/655668a42f9d4aa8bb469ed646344fe7 : SUCCESS in 1h 29m 29s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a0384bda51a42d9abd65dc01d29dd5d : SUCCESS in 1h 33m 08s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/165fe5e968204cd5b77b8491507be2dd : SUCCESS in 2h 58m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24c96b5c05089101cb5c9d4ed09474de8091c40c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-05-04 02:27:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"59e3c55a42e36f8f8b4cd7b15bd533c15d40fe45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 02:28:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab9a7ab52f4d771c354de7716ef461895499f0e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 03:53:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/43f41f5184494b77af8218fc0c0ff62f : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27bb53aaeea4df29357d130176422c7 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bba4ab66cb084d5d9f2422a0296b1e31 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa3b53869936482db8eef6d5a5fddb62 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04eb6265afb4609bf19deecd658cb65 : SUCCESS in 6m 51s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1703115b64124ae69e03a2ba50d35ee6 : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":1},{"id":"da23af36adcccbb6062726512d76894887f71bad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 03:53:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4a806037885c73908e70daa4d381aa609873f63d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 03:54:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a5499bbed5a4edfb75c8dcd908be061 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e232cb800c8924ece60beb1590f50b25bd9f9b3a","revisions":{"e232cb800c8924ece60beb1590f50b25bd9f9b3a":{"kind":"REWORK","_number":1,"created":"2021-05-03 01:32:58.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/16/789116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/16/789116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/789116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/789116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/16/789116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/16/789116/1"}}},"commit":{"parents":[{"commit":"d5497179d4789ef1c22ad8f9a39c719459241a62","subject":"Fix check_subnet_router filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d5497179d4789ef1c22ad8f9a39c719459241a62"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2021-05-03 01:32:24.000000000","tz":-300},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2021-05-03 01:32:57.000000000","tz":-300},"subject":"setup.cfg: add oslo.policy.enforcer","message":"setup.cfg: add oslo.policy.enforcer\n\nThis allows generation of sample policies, policy diffs, etc.\n\nChange-Id: I41d989d876a73f49684d8212d27e48876190f210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e232cb800c8924ece60beb1590f50b25bd9f9b3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/e232cb800c8924ece60beb1590f50b25bd9f9b3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
