)]}'
{"id":"openstack%2Ftrove~789415","triplet_id":"openstack%2Ftrove~master~I2f6acb879004feb4d935ba81e3dabf6e54ec2ab4","project":"openstack/trove","branch":"master","topic":"fix-periodic","hashtags":[],"change_id":"I2f6acb879004feb4d935ba81e3dabf6e54ec2ab4","subject":"Fix periodic public image job","status":"MERGED","created":"2021-05-03 20:28:44.000000000","updated":"2021-05-04 01:49:59.000000000","submitted":"2021-05-04 01:47:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789415-fix-periodic","meta_rev_id":"5f9d1485aef8be52467b86a9bd1160df73876e4a","_number":789415,"virtual_id_number":789415,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-04 01:47:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-05-04 00:22:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-05-04 00:22:16.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 22:37:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5c5b0af839270091ab2e8358ee3277dc2dc2b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-05-03 20:28:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e6c40c7473f8de87dc76cc1cf948a495549b034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 22:37:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14601f7dae2249d381781d86c9d125d7 : FAILURE in 8m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3fc0066fda43e5a121cf650de12443 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92aa5c05eff249e7be09c5ed96645ea3 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e552864d5362468292d01daa2d13a4a9 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ed19c90da114a13b8d1704c1f4223ab : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/195f5775e8c745d29d8a3cab2e888e0a : SUCCESS in 5m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7ad6ae3698514287a8c64cfc7a1af324 : SUCCESS in 6m 45s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/9411aa93272a4379afb93bbc5a10443e : SUCCESS in 1h 22m 47s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/d61b7b70acb54d9daf4083cd08f4df41 : SUCCESS in 1h 03m 54s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46f1614c96894387ba944a2ee109fa91 : SUCCESS in 2h 02m 31s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/43580a8a65c942a9bacc6c7d2dcce79b : FAILURE in 53m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b33903216e48d8f014335e4cdd93b09062d05f7c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-05-04 00:22:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ed32cd8924319596c20fe4967ea2294fa9e34f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 00:22:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3b9bd86e3b32e7fd461dd311ddc2d50f1e58b50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 01:47:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8189580f69ec4532b77f157ad879f340 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53592701458640249b614a22cdd2907f : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20a07e7cd25848559cc9acbd76cb62a7 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22bacba3f4ee456fb975de604e214dc9 : SUCCESS in 6m 34s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/81983608bf6c40d4b0fb2bcc74fa0fc8 : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3e39df00e5c9830879da3a0816db27d674a2c9f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 01:47:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5f9d1485aef8be52467b86a9bd1160df73876e4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 01:49:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6b7b2c682b45efaab45a0d2b50e777 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fd6e5f8003295da49d558b6f37dfdd5b573772c","revisions":{"1fd6e5f8003295da49d558b6f37dfdd5b573772c":{"kind":"REWORK","_number":1,"created":"2021-05-03 20:28:44.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/15/789415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/15/789415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/789415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/789415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/15/789415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/15/789415/1"}}},"commit":{"parents":[{"commit":"968f224a6614f3bf04b7eef9f7233601b2754ef6","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/968f224a6614f3bf04b7eef9f7233601b2754ef6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-03 20:28:35.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-03 20:28:35.000000000","tz":720},"subject":"Fix periodic public image job","message":"Fix periodic public image job\n\nChange-Id: I2f6acb879004feb4d935ba81e3dabf6e54ec2ab4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1fd6e5f8003295da49d558b6f37dfdd5b573772c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1fd6e5f8003295da49d558b6f37dfdd5b573772c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
