)]}'
{"id":"openstack%2Ftrove~789428","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~I6046f079db210c08ddef22ac2bbe564cb65b8e25","project":"openstack/trove","branch":"stable/victoria","topic":"victoria-fix-periodic","hashtags":[],"change_id":"I6046f079db210c08ddef22ac2bbe564cb65b8e25","subject":"Fix periodic public image job","status":"MERGED","created":"2021-05-04 00:25:11.000000000","updated":"2021-05-04 06:50:16.000000000","submitted":"2021-05-04 06:47:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789428-victoria-fix-periodic","meta_rev_id":"3261bb0bc1c3be44dba490144efea74264f58856","_number":789428,"virtual_id_number":789428,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-04 06:47:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-05-04 04:51:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-05-04 04:51:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 00:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"027f9e48695804c0e43d9a4f25129308c24c3b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-05-04 00:25:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31aa22eb00336a52f29367f7bf9998dec272f69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 00:25:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/trove on branch stable/victoria.  The error was:\n\n  expected str for dictionary value @ data[\u0027periodic\u0027][\u0027jobs\u0027][0][\u0027publish-trove-guest-image-ubuntu-bionic\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        # - openstack-lower-constraints-jobs\n        - openstack-python3-victoria-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-cover:\n              voting: false\n          - trove-tox-bandit-baseline:\n              voting: false\n          - trove-tempest:\n              voting: false\n          - trove-tempest-ipv6-only:\n              voting: false\n      gate:\n        queue: trove\n        jobs:\n          - trove-tempest:\n              voting: false\n      experimental:\n        jobs:\n          - trove-functional-mysql\n          - trove-scenario-mysql-single\n          - trove-scenario-mysql-multi\n          - trove-scenario-mariadb-single\n          - trove-scenario-mariadb-multi\n      periodic:\n        jobs:\n          - publish-trove-guest-image-ubuntu-bionic:\n          - publish-trove-guest-image-ubuntu-bionic-dev:\n\n  in \"openstack/trove/.zuul.yaml@stable/victoria\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"e0285555fa4db983ce43f8af14a0d60ba99d545a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-05-04 00:43:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"60f09f020b39ab70907f381970893aff157eaffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 02:56:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d1704666e9c4267a3c9713a616143c5 : FAILURE in 8m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/accb0617beab4d0cac16607860b6616d : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65bf11b3bc7640fea8bbd010a6434997 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dca739e7482844429c0a7fb52eac2de3 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe84f65bde5f4fe7aa62d93922c59ee8 : SUCCESS in 5m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d1786553e48d4e6da859cdf58b14ff0e : SUCCESS in 7m 07s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/2b1409e99b854827bed9f243621456c8 : FAILURE in 2h 08m 59s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/768863ad106749acbea0aa0db9c8b2af : TIMED_OUT in 2h 03m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"91b88cbe7eb0c9bdfcf4d8c08365700918b592ad","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-05-04 04:51:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"299ebe48cdf70b459a4cffe895137aac6063315f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 04:51:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d0956d49a1b3eead19ff0f097e27f5bbb783a2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 06:47:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c0d189df18400a80dc0f0d055763a5 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5b05e6a69454c8ba1836d82aff868eb : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae20868b45774df394d464454fbfa891 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50aaeb8aa1741ca8e658b37d4f8e6ce : SUCCESS in 5m 51s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/10d7deb86d4444c383e306e5c7d98f4c : FAILURE in 1h 54m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00bcc572defda2d0531652ee76ea61fc41116fcf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 06:47:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3261bb0bc1c3be44dba490144efea74264f58856","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 06:50:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ed2170a7334b31bb5f5e5a67ec7861 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5813fb205e5fe7cc51b8d55ca183a9b164a10af7","revisions":{"0982e8e48691a8dab591664e26c56be6ddce727d":{"kind":"REWORK","_number":1,"created":"2021-05-04 00:25:11.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/28/789428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/789428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/789428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/789428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/789428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/789428/1"}}},"commit":{"parents":[{"commit":"003b98a73e5da99eadcc1806f3e0cc9a109ea1cb","subject":"Fix check_subnet_router filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/003b98a73e5da99eadcc1806f3e0cc9a109ea1cb"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-04 00:24:37.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-04 00:24:37.000000000","tz":720},"subject":"Fix periodic public image job","message":"Fix periodic public image job\n\nCherry picked from\nhttps://review.opendev.org/c/openstack/trove/+/789415\n\nChange-Id: I6046f079db210c08ddef22ac2bbe564cb65b8e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0982e8e48691a8dab591664e26c56be6ddce727d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0982e8e48691a8dab591664e26c56be6ddce727d"}]},"branch":"refs/heads/stable/victoria"},"5813fb205e5fe7cc51b8d55ca183a9b164a10af7":{"kind":"REWORK","_number":2,"created":"2021-05-04 00:43:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/28/789428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/28/789428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/789428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/789428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/28/789428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/28/789428/2"}}},"commit":{"parents":[{"commit":"003b98a73e5da99eadcc1806f3e0cc9a109ea1cb","subject":"Fix check_subnet_router filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/003b98a73e5da99eadcc1806f3e0cc9a109ea1cb"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-04 00:24:37.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-04 00:43:08.000000000","tz":0},"subject":"Fix periodic public image job","message":"Fix periodic public image job\n\nCherry picked from\nhttps://review.opendev.org/c/openstack/trove/+/789415\n\nChange-Id: I6046f079db210c08ddef22ac2bbe564cb65b8e25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5813fb205e5fe7cc51b8d55ca183a9b164a10af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5813fb205e5fe7cc51b8d55ca183a9b164a10af7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
