)]}'
{"id":"openstack%2Ftrove~790639","triplet_id":"openstack%2Ftrove~master~I5acc467985d80fe4eb7d7a69055c9b178bcebbd1","project":"openstack/trove","branch":"master","topic":"setup-cfg","attention_set":{},"removed_from_attention_set":{"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2022-04-28 13:49:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5acc467985d80fe4eb7d7a69055c9b178bcebbd1","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-05-11 09:45:17.000000000","updated":"2022-04-28 13:50:38.000000000","submitted":"2022-04-28 13:49:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790639-setup-cfg","meta_rev_id":"8db2db4bd2b535b3c9f7c38acdfd79977d55f167","_number":790639,"virtual_id_number":790639,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-28 13:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"recommended":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"all":[{"value":1,"date":"2022-04-06 14:16:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-28 11:44:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-28 11:44:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 11:48:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 14:16:35.000000000","updated_by":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"reviewer":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"state":"REVIEWER"},{"updated":"2022-04-28 11:44:19.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"cf8bd23ffcf0be9dd35c23b2299b7a4bab0291bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-11 09:45:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10de37a1959f71eb49ea322ed3b7819cd60c8a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 11:48:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d4b45cda7ae143e6ab86398d1ca70663 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b85127d88ed5439d9c2240b819686a4a : FAILURE in 7m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1f20dbeefe04eb3a5f8bcb3c3ff82cc : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c72cc46980344f8cb243b9c7bc78f2df : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76ca1bd041ca462fa75598e6d1d4218f : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feccff075b814a4eb677b05eeb0965c8 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848ffe86ff7141d3966af54cb28840f9 : SUCCESS in 7m 04s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/774225c7f06540ec99a5679c298fc1b3 : SUCCESS in 1h 52m 34s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/acb644f26db94a4da3e1ba957ebfa07c : SUCCESS in 1h 08m 36s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8643b2295a1f453d8112f47aade94154 : FAILURE in 46m 41s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f7b058c20743496083729a1700b27a9c : SUCCESS in 2h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ecb3d62694291e1eb17c981a8d10f6b0621433e","author":{"_account_id":5572,"name":"Don Kehn","display_name":"DEKehn","email":"dekehn@gmail.com","username":"dekehn"},"date":"2022-04-06 14:16:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7bd9addffc26d800aa354ceef37296dae1c83dda","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-04-28 11:44:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"daeff9183ee8cdfb3e72d025a23ec7e7e8520158","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 11:44:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6764a0e9b38e6128dbdfd8ca2a8806472ebe0a7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 13:49:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa6fe508ee944243bfc7f7dea619b626 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52754d75e42a475997a73e64805d6942 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1226ab4e3e444769eebd8d1bb3dfb58 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9e441f7a3c144eca6e74142bb7a6181 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7066b48a7440c690fdd56d20373bda : SUCCESS in 6m 37s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/b8228e169aa5400196cc7a48ffca01b0 : FAILURE in 2h 03m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef0ce1b8c97202c9a20a7892b9f26b53a652d2eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 13:49:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8db2db4bd2b535b3c9f7c38acdfd79977d55f167","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 13:50:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93219a7998c494fb54586b3332ee2ca : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d6fa4fc72a1fb34a0d85108428a96f2cdf2ce87","revisions":{"5d6fa4fc72a1fb34a0d85108428a96f2cdf2ce87":{"kind":"REWORK","_number":1,"created":"2021-05-11 09:45:17.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/39/790639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/39/790639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/39/790639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/39/790639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/39/790639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/39/790639/1"}}},"commit":{"parents":[{"commit":"71d7b183193b8aa94b1eae4831a108b8c3534c30","subject":"Merge \"setup.cfg: add oslo.policy.enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d7b183193b8aa94b1eae4831a108b8c3534c30"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 09:44:51.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-11 09:44:51.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of\ndash-separated options in \u0027setup.cfg\u0027 will not be supported\nin a future version [1].\nGet ahead of the issue by replacing the dashes with underscores.\nWithout this, we see \u0027UserWarning\u0027 messages\nlike the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I5acc467985d80fe4eb7d7a69055c9b178bcebbd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5d6fa4fc72a1fb34a0d85108428a96f2cdf2ce87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5d6fa4fc72a1fb34a0d85108428a96f2cdf2ce87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
