)]}'
{"id":"openstack%2Ftrove~794090","triplet_id":"openstack%2Ftrove~master~Ia84885b36bb5a2ca81e0faec5e86a873bc5f6600","project":"openstack/trove","branch":"master","topic":"fix-wsgi-validation-error","attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-18 12:27:01.000000000","reason":"\u003cGERRIT_ACCOUNT_39300\u003e replied on the change","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}},"33480":{"account":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"last_update":"2026-09-18 10:37:14.000000000","reason":"\u003cGERRIT_ACCOUNT_39300\u003e replied on the change","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}},"6732":{"account":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"last_update":"2026-09-18 10:37:14.000000000","reason":"\u003cGERRIT_ACCOUNT_39300\u003e replied on the change","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}}},"removed_from_attention_set":{"39300":{"account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"last_update":"2026-09-18 12:27:01.000000000","reason":"\u003cGERRIT_ACCOUNT_39300\u003e replied on the change","reason_account":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}}},"hashtags":[],"change_id":"Ia84885b36bb5a2ca81e0faec5e86a873bc5f6600","subject":"Handle root-level API validation errors","status":"NEW","created":"2021-06-02 03:36:01.000000000","updated":"2026-09-18 13:39:22.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"120024a956231bbe4d5d0d1ad0cc2ca3ff394436","_number":794090,"virtual_id_number":794090,"owner":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-18 13:39:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}],"CC":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 06:05:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-08 03:33:42.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2026-09-18 10:36:37.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"CC"},{"updated":"2026-09-18 10:37:14.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"CC"},{"updated":"2026-09-18 10:37:14.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"22cda20ad76ddfad5a704e0d1ab2ecb51d860ac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-06-02 03:36:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54f529a9088ca6f02d011bffe28a2ece4a6bf375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 06:05:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a527a264b9c4af49dbb1681c1498ba5 : FAILURE in 7m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31109928b7bb424cb431b2884c33debc : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17c903bf961f401aaf796f0373537162 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9b995a40ccc49f2a24466b3555e3390 : SUCCESS in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b570d0acf4c4fcd984085713344aea0 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84801eba23ca4f24a897a77703ec4718 : SUCCESS in 6m 38s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/01c27d9e97344ac0813415ea1f0e94d2 : SUCCESS in 6m 01s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/79e0d04b22fd41cdafc71b89c869304a : FAILURE in 2h 28m 06s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/9479c1a55ea0484eb1fe5b7ddcd5b0d7 : SUCCESS in 1h 08m 55s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5082e9096e7444b4b03d888201143fb0 : FAILURE in 29m 55s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b56207128d944e23a1fbe84f824e1e53 : FAILURE in 1h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8bb5593df24cad19a5ff4968d8334d5b8a18680","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-08 03:33:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nPlease use story and task number for the issue reference, see https://docs.opendev.org/opendev/infra-manual/latest/developers.html#development-workflow","accounts_in_message":[],"_revision_number":1},{"id":"9b0f260362f60026819c0fce51723b8a7d7b7465","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-06-08 07:54:15.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Please use story and task number for the issue reference, see https://docs.opendev.org/opendev/infra-manual/latest/developers.html#development-workflow\n\nDone,please check again.","accounts_in_message":[],"_revision_number":1},{"id":"ba54b6b9b6c7bc4aa2b7ecf5bda40304f95d6bb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-06-15 02:31:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ebca861afce2a687505d24cc9573df829251e74c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-06-15 03:15:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad5cb8e0101cf6137ff0e140259eacbd9422040b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 05:22:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/454c9f559ae7451db03210dc6703f016 : FAILURE in 9m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293b63b1985c4603b9b900a0d5947e82 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8be7ac8f29f493e9add732b102f6d10 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de42f98856934d23914fbab8912c0989 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82f09533bb4e44d4b8869935276c8790 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d713b8c379a248b59c4bbb22dace591a : SUCCESS in 6m 11s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/6c3cc4422e904744848f3ab9f0d669d7 : SUCCESS in 6m 21s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/451b3da2a6ff42f48aa0e48be1d64116 : FAILURE in 2h 06m 42s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/228abca4cc93429393dbd405f54be9bb : SUCCESS in 1h 25m 51s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a55e90599c6464fb0eba3689fcc64ac : TIMED_OUT in 2h 02m 52s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/1d4f2c2542294f2a86418a4925221aae : FAILURE in 1h 07m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9037a01fd6e1f3a929218ac312c3dcb0f8756e24","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-18 08:26:23.000000000","message":"Patch Set 3: Code-Review-1\n\nDon\u0027t need to copy the whole story to the commit message. Please have some google for how to write good commit messages.\n\nAdditionally, please use the right story reference:\n\n```\nStory: 2008934\nTask: 42549\n```","accounts_in_message":[],"_revision_number":3},{"id":"157b20a60774b8061df6f224ca22911c090c3d83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-06-22 03:13:47.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9021dc5d1da7e55e5394d1bd5ce62a068bda038b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-06-22 03:17:23.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"42cd903767a6e447a9d4250b9714507c9d4075bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 05:26:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f59e634239d4a3d910dd09039141009 : FAILURE in 7m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c445b1ed6e54596b41d661d28a0a88a : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6f283af6cfb4790ba535a254d0f0880 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0757621672246239daa883f5d5148bd : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49f85348b83d4826840a3ca44b18bffb : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7b702516db441989e9b88475156232 : SUCCESS in 7m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/fce4f900276e430ab3170d1f0fb69702 : SUCCESS in 6m 14s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/134875713cfe4205aa73b61612b78b9e : FAILURE in 1h 32m 22s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/6ff0294252a9407d8aa135479ea2aa7c : SUCCESS in 1h 25m 04s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7969bd6231d743dd9866ccaf4b3f4570 : TIMED_OUT in 2h 03m 08s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2b7c7d3fc4cf4fcd8b5c6fecf78d3b13 : FAILURE in 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f8593f660d2df21f411cda3f81ba020c85c0e782","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-15 01:06:26.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"7db14d933d0301d53cec23236ce1ba8c270bc417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 03:12:13.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/157840f2551a44018018bdee1b7d52a4 : FAILURE in 9m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb36d1de72d444d1aeb8161fdec5fe4b : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bad784be963451da7e0f2dff8d77dee : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2193c6808ff9480c868c91272ad010ec : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b662ef6796b417ea783dbdfac5a98f0 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d2a02ec5284ade9a26fbaef8a4a419 : SUCCESS in 7m 58s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/053ff4d5e6d04319a1304a0119315c2d : SUCCESS in 6m 06s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/85ec678791084209a2ba355ec143a018 : FAILURE in 1h 11m 08s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/1cb694f895dc4f7585494fe15b16c429 : SUCCESS in 1h 17m 46s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/580ee0d552894f06b9b8b26051355056 : TIMED_OUT in 2h 04m 10s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/280240bd81114435b6316a038143aafb : FAILURE in 29m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"638be2fcb45b66b00c3630c30503888c7ebbcb4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-18 10:36:37.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"4b3c032d1ca5aee866c74da6a8bcf5c7b7e2dd6a","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-18 10:37:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5da147248bdcee3b2a58747900aea3ce6dae8b70","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 10:41:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6448a03f44a10cffd3174e16d85ca7fca773c301","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:52:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0d7fecad73a4cde9b90379c1d381eee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/313fbcfc1f594fbfbf463e1e894e683a : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fb8545a1ebfa40b7a5dc76a016d730c9 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/31714ce2cb99456b8f962ff8f948a1ab : SUCCESS in 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e7b96de3f889a339c2734f757799a14fe62dbd0f","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-18 11:05:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dcbf52d78b3259206ca91d31a1932342aa1b5c83","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 11:44:55.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ab009bc6691a9195bb55a8d738f72f3837e3afe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-18 12:20:36.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"23623c9d63567946f0b0fa9d1ae95914a4608394","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-18 12:27:01.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0d28216aee21cbf48b51249be2f6d684fc8af346","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 12:56:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3520b003d2be4394b4f06a4cf8a90827\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45684bb239b94eaf978f4907286fdbd6 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6e02294be23a4297866948f38d08386c : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/745e72966e5242b795409cbcec0b0081 : SUCCESS in 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"120024a956231bbe4d5d0d1ad0cc2ca3ff394436","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 13:39:22.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/696a77332d2c4d798c7745e825ca4d8e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3d5baedc1e34f3a968a0c56aa717f25 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58aab8ede95d45f79a4cda1807b9e6e6 : SUCCESS in 8m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35c0d71b670b4880a18fdd740cfa957f : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c294ff73a3dd4f30944725f58f172601 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b2d67ba31c9a457096f84d768d50374a : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d88287ec9b645b7bb96e2742e3c81b4 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6296da7159754865a4feb753e38ebdb3 : SUCCESS in 8m 40s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/a8fabafc363147e89732d9a90f8458c4 : SUCCESS in 52m 30s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/de0795005a974e3f86c178becdf3c817 : FAILURE in 21m 37s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/61f2085fa23745ec9f017ffc7dbf1de8 : SUCCESS in 1h 04m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/d8a1c026291c4b0293602df3b1b33a08 : SUCCESS in 1h 16m 29s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/a4f2341e8c0b4b3faeb904c6baf9dc26 : SUCCESS in 34m 48s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/db52bd5ac124480d9befafe8d5ec2a6d : SUCCESS in 1h 11m 21s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/802f90920ccd433f843d3eb9e471a2df : SUCCESS in 1h 13m 52s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/dbce3514cd5440be9b3953ad8dafb978 : SUCCESS in 1h 07m 34s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8fccf2bb801c4edc9ea1a0620f30b929 : SUCCESS in 34m 35s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/d55d4f200d6c4be3ac866bd105819a6b : SUCCESS in 1h 13m 30s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/d2b8a6b0d33a47f3b37fbca479bf0cf2 : SUCCESS in 51m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/ddbda6c9fd724a5a8d4ee95401345f1d : SUCCESS in 57m 21s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a6909231a794c5c8d26fd488a36e127 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"3a5a31e55ab3221939415c6b9bcbb33b3e149f68","revisions":{"4642b76cde75b143b4acc95f391ddc549cb7d494":{"kind":"REWORK","_number":1,"created":"2021-06-02 03:36:01.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/90/794090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/794090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/794090/1"}}},"commit":{"parents":[{"commit":"71d7b183193b8aa94b1eae4831a108b8c3534c30","subject":"Merge \"setup.cfg: add oslo.policy.enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/71d7b183193b8aa94b1eae4831a108b8c3534c30"}]}],"author":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-06-02 03:31:33.000000000","tz":480},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-06-02 03:31:33.000000000","tz":480},"subject":"BUG of trove/common/wsgi/Controller.format_validation_msg()","message":"BUG of trove/common/wsgi/Controller.format_validation_msg()\n\ntask: 42549\nstory: openstack-trove-bugfix-wocloud\nChange-Id: Ia84885b36bb5a2ca81e0faec5e86a873bc5f6600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4642b76cde75b143b4acc95f391ddc549cb7d494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4642b76cde75b143b4acc95f391ddc549cb7d494"}]},"branch":"refs/heads/master"},"f1b268435f4fe65b832fc4ca3a2ba734cb0c7522":{"kind":"REWORK","_number":2,"created":"2021-06-15 02:31:00.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/90/794090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/794090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/794090/2"}}},"commit":{"parents":[{"commit":"968f224a6614f3bf04b7eef9f7233601b2754ef6","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/968f224a6614f3bf04b7eef9f7233601b2754ef6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-03 20:28:35.000000000","tz":720},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-06-15 02:26:37.000000000","tz":480},"subject":"Error message from wsgi is not right","message":"Error message from wsgi is not right\n\nTrove version:\nrocky\n\nStep to reappear:\n[CMD]# curl -H \"Content-Type: application/json\" -H \"X-Auth-Token: $X_SUBJECT_TOKEN\" -X POST ${OS_TROVE_ENDPOINT_VERSIONS_URL}/v1.0/${QUERY_TENANT_ID}/instances -d \"{}\"\nWhen the request body is \"{}\",use curl to request the Trove-Instance-Create API.\n\nBefore Bug-fix:\nThe error type is not accurate:\n{\n    \"instanceFault\": {\n        \"code\": 500,\n        \"message\": \"Internal Server Error. Please keep this ID to help us figure out what went wrong: (a71dcbff-07e6-46dc-8453-b8657c498b72).\"\n    }\n}\n\nAfter Bug-fix:\nThe error type is right:\n{\n    \"badRequest\": {\n        \"code\": 400,\n        \"message\": \"Validation error: \u003croot (zero path length)\u003e \u0027instance\u0027 is a required property\"\n    }\n}\n\ntask: 42549\nstory: openstack-trove-bugfix-wocloud\nChange-Id: Ia84885b36bb5a2ca81e0faec5e86a873bc5f6600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f1b268435f4fe65b832fc4ca3a2ba734cb0c7522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f1b268435f4fe65b832fc4ca3a2ba734cb0c7522"}]},"branch":"refs/heads/master"},"d42519381410e66c7e79b893340d5585ce3337a1":{"kind":"REWORK","_number":3,"created":"2021-06-15 03:15:07.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/90/794090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/794090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/794090/3"}}},"commit":{"parents":[{"commit":"968f224a6614f3bf04b7eef9f7233601b2754ef6","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/968f224a6614f3bf04b7eef9f7233601b2754ef6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-03 20:28:35.000000000","tz":720},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-06-15 03:12:24.000000000","tz":480},"subject":"Error message from wsgi is not right","message":"Error message from wsgi is not right\n\nTrove version:\nrocky\n\nStep to reappear:\n[CMD]# curl -H \"Content-Type: application/json\" -H \"X-Auth-Token: $X_SUBJECT_TOKEN\" -X POST ${OS_TROVE_ENDPOINT_VERSIONS_URL}/v1.0/${QUERY_TENANT_ID}/instances -d \"{}\"\nWhen the request body is \"{}\",use curl to request the Trove-Instance-Create API.\n\nBefore Bug-fix:\nThe error type is not accurate:\n{\n    \"instanceFault\": {\n        \"code\": 500,\n        \"message\": \"Internal Server Error. Please keep this ID to help us figure out what went wrong: (a71dcbff-07e6-46dc-8453-b8657c498b72).\"\n    }\n}\n\nAfter Bug-fix:\nThe error type is right:\n{\n    \"badRequest\": {\n        \"code\": 400,\n        \"message\": \"Validation error: \u003croot (zero path length)\u003e \u0027instance\u0027 is a required property\"\n    }\n}\n\ntask: 42549\nstory: openstack-trove-bugfix-wocloud\nChange-Id: Ia84885b36bb5a2ca81e0faec5e86a873bc5f6600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d42519381410e66c7e79b893340d5585ce3337a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d42519381410e66c7e79b893340d5585ce3337a1"}]},"branch":"refs/heads/master"},"a39c3a975e35e5b2c6c186b4a067794f51d25e1f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-06-22 03:13:47.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/90/794090/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/794090/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/794090/4"}}},"commit":{"parents":[{"commit":"968f224a6614f3bf04b7eef9f7233601b2754ef6","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/968f224a6614f3bf04b7eef9f7233601b2754ef6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-03 20:28:35.000000000","tz":720},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-06-22 02:46:00.000000000","tz":480},"subject":"Error message from wsgi is not right","message":"Error message from wsgi is not right\n\nTrove version:\nrocky\n\nWhen the request body is \"{}\",use curl to request the Trove-Instance-Create API.\nFunction trove.format_validation_msg will go wrong because list \u0027path\u0027 is null so that path[0] is out of range.\nAS a result wsgi return a HTTP code 500(Internal Server Error) but actually wsgi should return HTTP code 400(badRequest)\n\ntask: 42549\nstory: 2008934\nChange-Id: Ia84885b36bb5a2ca81e0faec5e86a873bc5f6600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a39c3a975e35e5b2c6c186b4a067794f51d25e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a39c3a975e35e5b2c6c186b4a067794f51d25e1f"}]},"branch":"refs/heads/master"},"46d18f667323dc3177fbaf08beb4dea4ea75851f":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-06-22 03:17:23.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/90/794090/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/794090/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/794090/5"}}},"commit":{"parents":[{"commit":"968f224a6614f3bf04b7eef9f7233601b2754ef6","subject":"Merge \"Dropping lower constraints testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/968f224a6614f3bf04b7eef9f7233601b2754ef6"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-03 20:28:35.000000000","tz":720},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-06-22 03:14:20.000000000","tz":480},"subject":"Error message from wsgi is not right","message":"Error message from wsgi is not right\n\nTrove version:\nrocky\n\nWhen the request body is \"{}\",use curl to request the Trove-Instance-Create API.\nFunction trove.common.wsgi.format_validation_msg will go wrong because list \u0027path\u0027 is null so that path[0] is out of range.\nAS a result wsgi return a HTTP code 500(Internal Server Error) but actually wsgi should return HTTP code 400(badRequest)\n\ntask: 42549\nstory: 2008934\nChange-Id: Ia84885b36bb5a2ca81e0faec5e86a873bc5f6600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46d18f667323dc3177fbaf08beb4dea4ea75851f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/46d18f667323dc3177fbaf08beb4dea4ea75851f"}]},"branch":"refs/heads/master"},"a7bd3cc740c19bf35afcfa961371fa389e349270":{"kind":"REWORK","_number":6,"created":"2026-09-18 10:36:37.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/90/794090/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/794090/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/794090/6"}}},"commit":{"parents":[{"commit":"20fb23f5a5a2b76fc997fe28de52b38beff953a5","subject":"Merge \"docs: Fix DevStack install instructions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/20fb23f5a5a2b76fc997fe28de52b38beff953a5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-03 20:28:35.000000000","tz":720},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-18 10:35:41.000000000","tz":300},"subject":"Error message from wsgi is not right","message":"Error message from wsgi is not right\n\nWhen the request body is \"{}\",use curl to request the Trove-Instance-\nCreate API.\n\nFunction trove.common.wsgi.format_validation_msg will go wrong because\nlist \u0027path\u0027 is null so that path[0] is out of range. As a result wsgi\nreturn a HTTP code 500(Internal Server Error) but actually wsgi should\nreturn HTTP code 400(badRequest)\n\ntask: 42549\nstory: 2008934\nChange-Id: Ia84885b36bb5a2ca81e0faec5e86a873bc5f6600\nSigned-off-by: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nCo-authored-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a7bd3cc740c19bf35afcfa961371fa389e349270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/a7bd3cc740c19bf35afcfa961371fa389e349270"}]},"branch":"refs/heads/master"},"3a5a31e55ab3221939415c6b9bcbb33b3e149f68":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-09-18 12:20:36.000000000","uploader":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"ref":"refs/changes/90/794090/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/90/794090/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/90/794090/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/90/794090/7"}}},"commit":{"parents":[{"commit":"20fb23f5a5a2b76fc997fe28de52b38beff953a5","subject":"Merge \"docs: Fix DevStack install instructions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/20fb23f5a5a2b76fc997fe28de52b38beff953a5"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-05-03 20:28:35.000000000","tz":720},"committer":{"name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","date":"2026-09-18 12:17:53.000000000","tz":300},"subject":"Handle root-level API validation errors","message":"Handle root-level API validation errors\n\njsonschema returns an empty error.path for validation errors at the\ndocument root. format_validation_msg accesses the first path element\nunconditionally, which raises IndexError and returns HTTP 500.\n\nHandle an empty error path so these requests return HTTP 400 with the\nvalidation message.\n\nStory: 2008934\nTask: 42549\nChange-Id: Ia84885b36bb5a2ca81e0faec5e86a873bc5f6600\nSigned-off-by: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nCo-authored-by: Olessya Khussainova \u003colessya.khussainova@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3a5a31e55ab3221939415c6b9bcbb33b3e149f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/3a5a31e55ab3221939415c6b9bcbb33b3e149f68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
