)]}'
{"id":"openstack%2Ftrove~796550","triplet_id":"openstack%2Ftrove~master~I98c9c1d193934c37e77297a04231a24585674306","project":"openstack/trove","branch":"master","topic":"fix-sqlalchemy","hashtags":[],"change_id":"I98c9c1d193934c37e77297a04231a24585674306","subject":"Fix sqlalchemy engine listener","status":"MERGED","created":"2021-06-15 23:33:57.000000000","updated":"2021-06-16 13:03:07.000000000","submitted":"2021-06-16 13:01:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796550-fix-sqlalchemy","meta_rev_id":"17b1a209cb3c43e4af469da5fcbac366c7b2d3f7","_number":796550,"virtual_id_number":796550,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 13:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-06-16 04:19:49.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-06-16 04:19:49.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-16 02:04:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6175bb5c857ec89baaef3a8724d2e68ed2f72056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-15 23:33:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91682cea136fa3e9d99bc117795b8dafec3e92aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-15 23:34:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9d80e91ecb1e65329acc3213379026136c95c193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 02:04:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/559a4b44e1844bf08e8ff6e95192700c : FAILURE in 7m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8d1d6524ed4fcbb5b9d55cce1cba36 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c025fc4b930541ee86ac485bbc7bd59e : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a08010640cb430d8935e502e5f87a88 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22d30d9f01194bf3bfa15c6b25c89551 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/378ddcbcd7564715a8e6e7af3d984934 : SUCCESS in 6m 14s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a28e4087ba5c46d8814070ee13b08707 : SUCCESS in 6m 43s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3fe79b43dffb432cba6f053d172c7852 : SUCCESS in 1h 19m 57s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/836f80470a124b2ea8ef5aca4d5977da : SUCCESS in 1h 14m 18s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbb554af04814fcfac62dd562153c237 : TIMED_OUT in 2h 02m 50s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/a17d3d585bde448f8af1a1231ad8a68a : SUCCESS in 2h 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"406100962ca2c4704d7e6a68623a4353b1a2da85","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-16 04:19:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5ece5e720016bb9464d8b8b002b89ff445f58667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 04:20:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b61b1c7436a9dbd8d1e9af14c61a03a1e9449785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 04:51:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1218750982be4de7a5925b66f3b5b7ff : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c83138e1f6e44daea208c405a7b24a79 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7617800f19fa47c29df14a310ff2a3d3 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d6b59fd910742e6aa090fb26fc1f5be : SUCCESS in 6m 06s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ff6328b8ebc94f5ab0769b974d2a0778 : FAILURE in 30m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1057c88a5c4fb15e71e87b7d832aa7a22a75d3f9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-16 09:51:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d9a3f223e9e016541f4ed2989deed1adc35d3ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:58:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e6f2b319b0d4ed78711877e38672251 : FAILURE in 6m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae0296c70ab484abe2da36c6689581b : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/636a955f7645415da5a998cd66465a89 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75dc2834e3dc4066923071d0cac34af3 : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5c360bbe34a4e4c94ce68a07c6f8fa7 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d57ee210a14368a55b72e7864b0a89 : SUCCESS in 6m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0d7ab6b7bf7741fcb7ea67282ed819cd : SUCCESS in 7m 48s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/46ee879cc8004ae3857683ddf172c019 : SUCCESS in 1h 43m 11s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c1078e4d24284f4fb3adf0678eeffb4a : SUCCESS in 1h 39m 43s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c06b0a4773674455bc08fd3617aa5c9c : TIMED_OUT in 2h 02m 51s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/564dc14bb60c4a909afeb568261dc44d : SUCCESS in 1h 50m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34072a0860164d605f9a90d4feef0b95424911c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:58:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a079141cf016cd7d4a3fa038d8862888c4d791a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:01:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594d426d10bf42a7b508ba794a127977 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0654325394cc48288ae7f3e3b567d4d3 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2fb9036824642d889e3fc1e86018a34 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/621ebc401ede4cf3adcacc7a84627a83 : SUCCESS in 6m 30s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/065c677379a14882a779e3fdf33667c3 : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":2},{"id":"5ed3adc32b08ecda61cbcf5f3d642e9d34e287a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:01:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"17b1a209cb3c43e4af469da5fcbac366c7b2d3f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:03:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4746b5bd1f944d51ab1d9a2283404e22 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8f91898de992ca63511d124b5382ad3b542a9969","revisions":{"0bfe08d6f33665b89835ee7185ddebf83906664f":{"kind":"REWORK","_number":1,"created":"2021-06-15 23:33:57.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/50/796550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/50/796550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/796550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/796550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/796550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/50/796550/1"}}},"commit":{"parents":[{"commit":"7ebce17e463e07d29dbc4f09669ca825c1f15b0d","subject":"Merge \"Changed minversion in tox to 3.18.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebce17e463e07d29dbc4f09669ca825c1f15b0d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-12 11:30:10.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-15 23:27:20.000000000","tz":720},"subject":"Fix sqlalchemy engine listener","message":"Fix sqlalchemy engine listener\n\nChange-Id: I98c9c1d193934c37e77297a04231a24585674306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0bfe08d6f33665b89835ee7185ddebf83906664f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/0bfe08d6f33665b89835ee7185ddebf83906664f"}]},"branch":"refs/heads/master"},"8f91898de992ca63511d124b5382ad3b542a9969":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-15 23:34:56.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/50/796550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/50/796550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/796550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/796550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/50/796550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/50/796550/2"}}},"commit":{"parents":[{"commit":"7ebce17e463e07d29dbc4f09669ca825c1f15b0d","subject":"Merge \"Changed minversion in tox to 3.18.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7ebce17e463e07d29dbc4f09669ca825c1f15b0d"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-12 11:30:10.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-15 23:34:01.000000000","tz":720},"subject":"Fix sqlalchemy engine listener","message":"Fix sqlalchemy engine listener\n\nThe sqlalchemy 1.4 had introduced some incompatible changes.\n\nChange-Id: I98c9c1d193934c37e77297a04231a24585674306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8f91898de992ca63511d124b5382ad3b542a9969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8f91898de992ca63511d124b5382ad3b542a9969"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
