)]}'
{"id":"openstack%2Ftrove~796988","triplet_id":"openstack%2Ftrove~master~Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6","project":"openstack/trove","branch":"master","topic":"bridge-network","hashtags":[],"change_id":"Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6","subject":"Use bridge network for db container","status":"MERGED","created":"2021-06-18 05:17:04.000000000","updated":"2021-06-22 10:55:54.000000000","submitted":"2021-06-22 10:54:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796988-bridge-network","meta_rev_id":"64ad9d582da7373add0e436eb39a122974ac5bf4","_number":796988,"virtual_id_number":796988,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-22 10:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-22 09:22:13.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-22 09:22:13.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 07:23:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04c35d3a189733dab61fdec395df278d0ac25fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-18 05:17:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a1791cb56b0877c4982aa2e775d07ac56f36d5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 07:23:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8fad67a3725438f84753980c5310bd8 : FAILURE in 13m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788f444746714f049933e44b93743586 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11c7aadfe1354409b25271781e2bd391 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85e62f51db18439e83178b63f5e059db : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b409224960944858933c3556160b644b : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b5dde5e2e64dcfb103f29dd093f680 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71ce5b107bdd4ef28fdbaf66de3db0e9 : FAILURE in 5m 14s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/33e694b7b26e45fd8aa36bc7ab51b7be : SUCCESS in 6m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/153b45246a784afe82b498a3e11d2a93 : FAILURE in 53m 09s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/25ba841887d64e69a73e8839f1629568 : FAILURE in 43m 10s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/822d78cd037948b481d3ec43d999ae51 : TIMED_OUT in 2h 04m 33s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/97d24cfddad64bdfa665518f04f31d05 : FAILURE in 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3d0b16df70539a9ae6c0e32005cb78f1856d77a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-18 10:26:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c691bd6845b685e6b11c11340302133534e97830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-18 11:17:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0e9d458061dd501249df3acdc155ee3a1b060d59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 13:25:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d810eeaeb3aa494a8769355ee8e1d4b3 : FAILURE in 6m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b51c7234b00f43239713d2d9652e5d84 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d025b71409a34601a68d5fb300549eee : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/863c682f24d04082a6277ae93c7e9880 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/624906d3be2c4f08a4ebbb874995499b : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/626489d14cd04f9c8ccd94f17d203111 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0032609a10b8450e8f8c371c479ab25d : FAILURE in 6m 07s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/8268d6373dcb464f966331e59be9c9c3 : SUCCESS in 7m 26s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/62f8f21a4d684a629ac7ed1ec3e9e60b : SUCCESS in 1h 05m 53s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/b2925be5253d4b108160dc82ccef5f0b : FAILURE in 54m 55s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/219c1cf3a8704791bb61c80d25282711 : TIMED_OUT in 2h 05m 33s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/4ae8f7ae5e4d4217b7f3698b129516e5 : FAILURE in 35m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7784de81007f536c01ecb7be032edc1a7038e65c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-19 11:33:19.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"36b21bc015e9091caca783ba1863179861fe9c0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-19 13:38:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37add8d7a0014c5b8f88d2ba98b728a0 : FAILURE in 6m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e08c7b906ae44f85af264361d5743bd9 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55abe7e781a941fbb192383d2529a826 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8caa330880da4f6caa6568f3405f0f6a : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f024f714819a46868f64ec441b383541 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a990440d5c394cce85b7e84c641297f6 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f13b5fb972964bf3b6532aa6e220cbdc : FAILURE in 5m 28s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/59306dea83034f92849522f8295960c8 : SUCCESS in 5m 31s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/47d2baeb267c4a799ad9905ae1392cb0 : SUCCESS in 1h 23m 53s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/fbc76bbd9881488cb7164386e5de9a44 : FAILURE in 1h 04m 21s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41f6491702fe4e71adf324a939ea2ef9 : TIMED_OUT in 2h 03m 31s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e667deb208824a05a642ac7238e04276 : FAILURE in 29m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"84db6580ec7c4f6796fc73743e73d2e533aec9af","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-20 09:06:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fc2e6145d4a273d6d72267825a5c76fca8cd96eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 11:15:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef572dbaf0a64ae2948a688f50f84258 : FAILURE in 6m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d88c5c4aadb4500ab098ac9ae363361 : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47a91abd343a4b03ae57fd7c54994c14 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/258c4691e5bf4663ad47b2ffeeac410d : SUCCESS in 7m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44a184b75b2b4a169797569be3bf7d22 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776df21686174ac28afc6eb957df65bc : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0cf02d69a37474bbc718ee01445a6d9 : FAILURE in 5m 12s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7a80dcb00b1c4969aab38732059a563a : SUCCESS in 5m 59s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1e7b736a3c5f49fab965e97cb8e35636 : SUCCESS in 1h 51m 05s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/f4a47974a0ed44a2abe4b9f8cb469a0e : FAILURE in 1h 21m 27s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/742921c843974651ad616af4dfccd4b4 : TIMED_OUT in 2h 02m 45s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/2767422d83d54360a0d3d413dd73bd5a : FAILURE in 37m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"72fe04989fd41e277537f60c872b4594ab7dd2c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-20 21:11:07.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"21722c897c33fcbf530b11286c50aa59c8ab3a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 23:17:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b73ec2fe514947bb865a44f2b879df55 : FAILURE in 7m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ba3a7557a6482d98ba9e2101d85644 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20e867d948aa4e3b927ef11cbc08ac2d : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8147929278c948999b9e1a5140c44ffc : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc4d4208c92341188b1bc1c9fa565709 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f78b237c82294d19b976911b833dfc6c : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45ae7a89b69f447085fba68637efbbb2 : FAILURE in 5m 42s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9774423044d41618ab593347ee48eeb : SUCCESS in 6m 49s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/1778e62fc79c43608fe6d4c2a1abc1ad : SUCCESS in 1h 15m 57s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/c832fe588a3c4b97bfb6fc6469ed15c4 : FAILURE in 1h 07m 55s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e4db4b0c2b3460aa94130a16122a81c : TIMED_OUT in 2h 03m 00s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3914a45a744f4879b15b099196f5cde3 : FAILURE in 52m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8e45d05ffe95f09a373cefb883a70e57c7bbd6c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-21 05:09:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"abb011233434974c530ddb735cc4cf6144655416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 07:17:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a1c719051ab481b847223ac3379df27 : FAILURE in 7m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0bf3c665ca942e0aa06a66cdf8d83d5 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc93eb24500748ef94d4fb671e28704a : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26229a79519646588cf6f1826ebdecab : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d36e3f9507a846e688085b7a032c60f0 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1dd45cfaad4bb5bc156300180731e6 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6d1f980b9dc4a149a571e1d0e219b52 : FAILURE in 6m 48s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/de9484cd73fe42878d07a1c69300b0dd : SUCCESS in 6m 44s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/c499afefc91d427e8af339e9ea54cadf : SUCCESS in 1h 22m 37s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/3d8baa7796cf41f1a33630052bfcd305 : FAILURE in 1h 13m 16s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a019747a2b58432288150a0cb277c670 : TIMED_OUT in 2h 03m 20s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ee08505ec56b4f9eb1954c7174b25356 : FAILURE in 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7dc50be1206109af87abd2a6a80ca0a1b77f1885","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-21 23:49:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a31090300e5f1b59d2879d0cce0f0d50dce50a05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 01:53:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db3105134ab247608c38a93306bfee47 : FAILURE in 7m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/716cc0e501ec4dc9aff1747095042f4e : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc98f7f43522484e91f2ae8a60ceeecb : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7da44e74d1f94ec59fdc46455f2807da : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/410ed1ab6f8f4ee7953b7b782f027a90 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a06d6508813c48a69296f72459adf3bd : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c5b28d5a5864d9fb1319d02c1102b85 : FAILURE in 5m 17s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/77df086465cf4508ae3022c2c40d22ff : SUCCESS in 6m 12s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3eefa60364e54fe3a9e1ad5e2dc10be2 : SUCCESS in 1h 33m 32s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/3f8643903b604ffd8cddc05566be5526 : SUCCESS in 1h 23m 09s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebab9f9e5e914d2682a11202d4ae86aa : TIMED_OUT in 2h 03m 30s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/48a00432f9dd4752b2278dbe1f5d65e1 : FAILURE in 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9863dd90dcf189ab40864805406f44b1321b65d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-22 04:25:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c45bc199b8ab7f08ffb86da0b265649786568762","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-22 04:33:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"56b100e99c75c46044e80fc21c9d9aa792f7c7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 06:39:16.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/988891758e014114b5dafccc9e9d56f3 : FAILURE in 9m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a17680afa94da6b015986315075387 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/557e31ec614d4fd5ba4643e212fce929 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79b5c9e488d14327b213cf2d9cce8807 : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ac33a4580b24a889875cc75bd887a11 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/258b68a2d6b54f6bb5964210afed937c : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96b297b678e94d589fa9437e10847d33 : FAILURE in 6m 10s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a9a8b1ce17aa4149bf0bca3b96b53dfc : SUCCESS in 6m 38s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e59537f2f1bf4cf1a8078336554d280f : SUCCESS in 1h 23m 26s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/bec9d9f6a13145b3aae5b250322c8df9 : SUCCESS in 1h 14m 46s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16cc39c420ae43bc9819f8cc9619dac9 : TIMED_OUT in 2h 04m 38s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/169df6ce9098416aa5f2f0ec60817e62 : FAILURE in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9e9796da7afafa66976f084b3d42bda2705681e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-22 07:10:21.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7532a2db5aa34ae282ea6c7a1a8ca416a61a0e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 09:17:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6513dee884c46999c9675e281c23f6e : FAILURE in 7m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd7256e94b54cbaa7d9eb04073d27c3 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e24a219df6194ed3a5b83749fd8f599b : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d2fb223029f4692a48a79a7f016149f : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63415ae1f17844589a9811bb2f6343d7 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02eb9a1acb3847c68956b6b267027ec1 : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cce4df51cde5443e9b12683af3d27362 : SUCCESS in 7m 31s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e362bae6c28f4da4a2501ff7219754a6 : SUCCESS in 9m 30s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ee5602f578904acab6b20218de32b8dc : SUCCESS in 1h 07m 35s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/5dcca5b0b9d1479db4007cac12b45d6f : SUCCESS in 1h 13m 45s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77752c1104914e8e9afe7c6614c7e8d2 : TIMED_OUT in 2h 05m 11s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/98f3d2a4643a4ea9b2ed0f8aaad9446a : FAILURE in 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b085388f5ced76c8109aa84a99c056da92475ca1","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-22 09:22:13.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"eb1ffa5b1803ab51334fa5f3eb997701a1980d12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 09:22:33.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"8c592b12ffde88429d98505bca3479ff149159d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 10:54:15.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0348668362c4448ac41b7443717cb5c : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9ec6f1d36d8497fa562a9e4ce959c30 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34c1bdbed52f47fe900b688306b1494a : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c76d1692701454e9324357077c088ee : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0aba2e502d744018a9b2c4adc7ff8e75 : SUCCESS in 8m 23s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/42a88e977f9440f49bef8d480c77decf : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":10},{"id":"2b28e6d062925772fc10ea77f41ba304178b4e75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 10:54:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"64ad9d582da7373add0e436eb39a122974ac5bf4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 10:55:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7366b36c64b4073a7146a7605b9a7af : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d32c42ba2da487d87ac57a03eff7b2a : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2","revisions":{"6d581cd945bac0694890514aca9829831c633ca5":{"kind":"REWORK","_number":1,"created":"2021-06-18 05:17:04.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/1"}}},"commit":{"parents":[{"commit":"2114e8a7750f1e6ed0aa2fbc36c7655abf48df0e","subject":"Enable tls-proxy support in test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2114e8a7750f1e6ed0aa2fbc36c7655abf48df0e"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-18 05:16:04.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\nChanged the network mode of database container to \"bridge\" and exposed\nthe service ports.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d581cd945bac0694890514aca9829831c633ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d581cd945bac0694890514aca9829831c633ca5"}]},"branch":"refs/heads/master"},"d37610f1c80e86a6e2209694eb317092bf4d69a2":{"kind":"REWORK","_number":2,"created":"2021-06-18 10:26:17.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/2"}}},"commit":{"parents":[{"commit":"2114e8a7750f1e6ed0aa2fbc36c7655abf48df0e","subject":"Enable tls-proxy support in test jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/2114e8a7750f1e6ed0aa2fbc36c7655abf48df0e"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-18 10:17:48.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\nChanged the network mode of database container to \"bridge\" and exposed\nthe service ports.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d37610f1c80e86a6e2209694eb317092bf4d69a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/d37610f1c80e86a6e2209694eb317092bf4d69a2"}]},"branch":"refs/heads/master"},"5ed25e928cf4ba146f9666e7c6a1012f6db0c241":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-18 11:17:19.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/3"}}},"commit":{"parents":[{"commit":"78a9f32e67064298c343cddcae18f32159fd7455","subject":"Support customized database container images in DevStack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/78a9f32e67064298c343cddcae18f32159fd7455"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-18 11:17:08.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\nChanged the network mode of database container to \"bridge\" and exposed\nthe service ports.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5ed25e928cf4ba146f9666e7c6a1012f6db0c241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/5ed25e928cf4ba146f9666e7c6a1012f6db0c241"}]},"branch":"refs/heads/master"},"7be851d90d459bd3db539aaaf83e656b7dfb5bfd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-19 11:33:19.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/4"}}},"commit":{"parents":[{"commit":"03e5512695f95aae4822c78db9f4cf9ec6eef7f9","subject":"Support customized database container images in DevStack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/03e5512695f95aae4822c78db9f4cf9ec6eef7f9"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-19 11:33:08.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\nChanged the network mode of database container to \"bridge\" and exposed\nthe service ports.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7be851d90d459bd3db539aaaf83e656b7dfb5bfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7be851d90d459bd3db539aaaf83e656b7dfb5bfd"}]},"branch":"refs/heads/master"},"b9d508322067d12898e2e3f26d5ec68c5fc22bda":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-20 21:11:07.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/5"}}},"commit":{"parents":[{"commit":"6d2ab68a8aa0406ee6320e778b8e290827c9e730","subject":"Fix backup using customized container image registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d2ab68a8aa0406ee6320e778b8e290827c9e730"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-20 21:07:00.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\nChanged the network mode of database container to \"bridge\" and exposed\nthe service ports.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9d508322067d12898e2e3f26d5ec68c5fc22bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9d508322067d12898e2e3f26d5ec68c5fc22bda"}]},"branch":"refs/heads/master"},"173fb533a2ec4ec9c195948bf4730933071fac09":{"kind":"REWORK","_number":6,"created":"2021-06-21 05:09:40.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/6"}}},"commit":{"parents":[{"commit":"6d2ab68a8aa0406ee6320e778b8e290827c9e730","subject":"Fix backup using customized container image registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d2ab68a8aa0406ee6320e778b8e290827c9e730"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-21 04:50:06.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\nChanged the network mode of database container to \"bridge\" and exposed\nthe service ports.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/173fb533a2ec4ec9c195948bf4730933071fac09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/173fb533a2ec4ec9c195948bf4730933071fac09"}]},"branch":"refs/heads/master"},"7dd44025fc179a5544929b7c7875855509d14001":{"kind":"REWORK","_number":7,"created":"2021-06-21 23:49:58.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/7"}}},"commit":{"parents":[{"commit":"6d2ab68a8aa0406ee6320e778b8e290827c9e730","subject":"Fix backup using customized container image registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d2ab68a8aa0406ee6320e778b8e290827c9e730"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-21 23:42:38.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\nChanged the network mode of database container to \"bridge\" and exposed\nthe service ports.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7dd44025fc179a5544929b7c7875855509d14001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7dd44025fc179a5544929b7c7875855509d14001"}]},"branch":"refs/heads/master"},"55a52aa71ae92c64699a584e675e2844d9ae58f2":{"kind":"REWORK","_number":8,"created":"2021-06-22 04:25:50.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/8"}}},"commit":{"parents":[{"commit":"6d2ab68a8aa0406ee6320e778b8e290827c9e730","subject":"Fix backup using customized container image registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d2ab68a8aa0406ee6320e778b8e290827c9e730"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-22 04:11:37.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\nChanged the network mode of database container to \"bridge\" and exposed\nthe service ports.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/55a52aa71ae92c64699a584e675e2844d9ae58f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/55a52aa71ae92c64699a584e675e2844d9ae58f2"}]},"branch":"refs/heads/master"},"95347f102cc40d1a2d56f5bc8545c199248770bd":{"kind":"REWORK","_number":9,"created":"2021-06-22 04:33:12.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/9"}}},"commit":{"parents":[{"commit":"6d2ab68a8aa0406ee6320e778b8e290827c9e730","subject":"Fix backup using customized container image registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d2ab68a8aa0406ee6320e778b8e290827c9e730"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-22 04:30:39.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\n- Changed the network mode of database container to \"bridge\" and exposed\n  the service ports.\n- Use socket file to connect with the database.\n- Upgrade the backup container image for postgressql.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95347f102cc40d1a2d56f5bc8545c199248770bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/95347f102cc40d1a2d56f5bc8545c199248770bd"}]},"branch":"refs/heads/master"},"b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2":{"kind":"REWORK","_number":10,"created":"2021-06-22 07:10:21.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/88/796988/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/88/796988/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/88/796988/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/88/796988/10"}}},"commit":{"parents":[{"commit":"6d2ab68a8aa0406ee6320e778b8e290827c9e730","subject":"Fix backup using customized container image registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/6d2ab68a8aa0406ee6320e778b8e290827c9e730"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-17 23:30:03.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-22 07:10:14.000000000","tz":720},"subject":"Use bridge network for db container","message":"Use bridge network for db container\n\n- Changed the network mode of database container to \"bridge\" and exposed\n  the service ports.\n- Use socket file to connect with the database.\n- Upgrade the backup container image for postgressql.\n\nChange-Id: Id5b119f8a474befc3a2cd6e061bbffc4ae5f7bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
