)]}'
{"id":"openstack%2Ftrove~797307","triplet_id":"openstack%2Ftrove~stable%2Fwallaby~Iae4d8fc52a50da4fd347270271f6761ce766d5c3","project":"openstack/trove","branch":"stable/wallaby","topic":"story/2008917-stable/wallaby","hashtags":[],"change_id":"Iae4d8fc52a50da4fd347270271f6761ce766d5c3","subject":"Add float types to load_items to support configuration parameters of float type","status":"MERGED","created":"2021-06-21 23:56:18.000000000","updated":"2021-06-22 04:41:32.000000000","submitted":"2021-06-22 04:40:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797307-story/2008917-stable/wallaby","cherry_pick_of_change":792871,"cherry_pick_of_patch_set":3,"meta_rev_id":"2a75232263014425d827ef970bf25a17f32a4a2a","_number":797307,"virtual_id_number":797307,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-22 04:40:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-06-22 03:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-06-22 03:08:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"}],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-21 23:56:18.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":31525,"name":"chimeng","email":"chisimon98k@gmail.com","username":"MengChi"},"state":"REVIEWER"},{"updated":"2021-06-22 03:01:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f23b6abe7a233763df17106bf00177ae2d8b57bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-21 23:56:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c1ef342e1dfa9382ab7f9e0452291a98c57ff8d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 03:01:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c79e9004b3a4f40936f490b7069ff0e : FAILURE in 7m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95df30cde2ee4faa8027de66df984194 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d02b9f4173848babd2b7caf028b0416 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38553f6b851a45d6a5cccd372069f861 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ec8c4e9133a4b008dabee9b88103194 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5e6b0b6d6a416b88174d2e63b20810 : SUCCESS in 6m 32s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/a1a29dd155cb4c1798ba2efc8a34baa7 : SUCCESS in 6m 49s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bdb2c0c738584b7d9c8b25b2492e58b5 : SUCCESS in 1h 28m 25s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07dc9ea129cf46e1a4a169e11d414711 : TIMED_OUT in 2h 03m 18s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/13c9db91a14d456fb9159e7b80f043fe : TIMED_OUT in 3h 03m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11ebf35921b146c2e1fcd4d888b686c0934b6947","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-22 03:08:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64e4d2cf8839004a8a97a126b6ca102c727c8fa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 03:08:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a299f026e2704b410894c7f35b1cbc044a2aa21f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 04:40:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206bf97adddf4dd5a980c7d8f2b98ef5 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/864ad8c818ec491183856197f92f9892 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ac52d5cda344597b9382909d07dfc98 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78aa600e88a84bdc92ccf813cc37afd5 : SUCCESS in 7m 05s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d66c03e71bf44d1e8b9f2b602ee8177f : SUCCESS in 1h 30m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8daff8660d69996cea8ed614bd1b3da75e8cbf2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 04:40:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a75232263014425d827ef970bf25a17f32a4a2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 04:41:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3b2cc6e86140a1b5ae248e87c1225b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7bbcf43d272a328a9ad685918375a3c2049a0ad","revisions":{"c7bbcf43d272a328a9ad685918375a3c2049a0ad":{"kind":"REWORK","_number":1,"created":"2021-06-21 23:56:18.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/07/797307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/07/797307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/797307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/797307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/07/797307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/07/797307/1"}}},"commit":{"parents":[{"commit":"698c715af847b453c9595173c4601dd14ef57fb4","subject":"Fix periodic public image job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/698c715af847b453c9595173c4601dd14ef57fb4"}]}],"author":{"name":"chimeng","email":"chisimon98k@gmail.com","date":"2021-05-25 05:14:47.000000000","tz":480},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-21 23:56:18.000000000","tz":0},"subject":"Add float types to load_items to support configuration parameters of float type","message":"Add float types to load_items to support configuration parameters of float type\n\nComplement missing float types when loading configuration group parameters\nso that parameters of type float can be attached into the trove instance successfully\n\nStory: 2008917\nTask: 42508\nChange-Id: Iae4d8fc52a50da4fd347270271f6761ce766d5c3\n(cherry picked from commit 189d3493a844231dfa10062b5e632a3e931e50f2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c7bbcf43d272a328a9ad685918375a3c2049a0ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c7bbcf43d272a328a9ad685918375a3c2049a0ad"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
