)]}'
{"id":"openstack%2Ftrove~797308","triplet_id":"openstack%2Ftrove~stable%2Fwallaby~Ib35743e1c315bb7fd4d0fbd99eaa409a4bf5c792","project":"openstack/trove","branch":"stable/wallaby","topic":"bridge-network-stable/wallaby","hashtags":[],"change_id":"Ib35743e1c315bb7fd4d0fbd99eaa409a4bf5c792","subject":"Fix backup using customized container image registry","status":"MERGED","created":"2021-06-21 23:58:37.000000000","updated":"2021-06-22 04:42:34.000000000","submitted":"2021-06-22 04:40:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797308-bridge-network-stable/wallaby","cherry_pick_of_change":797218,"cherry_pick_of_patch_set":1,"meta_rev_id":"06988cf281c25aac2bdf13b556fcd0eff09f662d","_number":797308,"virtual_id_number":797308,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-22 04:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-06-22 03:08:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-06-22 03:08:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-22 03:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1762312c50f9dabe63c2ee0e63d8a17272e5e4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-21 23:58:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"47a37905695798d9e70adb3e8fae3022b698c5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 03:04:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/837b225a5f1a4d758d93ab9518f86b09 : FAILURE in 7m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f8478b934b940e68f87bd5dbfd103d7 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b9587b4dd9c40b2a1ce546f1627bfa9 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15d4704d8668451ab172928da4f2e66b : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdcafbe6e03d4498b191f3f6ddf6da5c : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ab234510624db7803b2bd22911a195 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95b5ec1f5db74e46a15e591472fca950 : SUCCESS in 7m 24s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1580e255b0574b609e841d7c517ba420 : SUCCESS in 6m 37s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/308a0eb8ebc046f283cf4a3f78d88d52 : SUCCESS in 1h 25m 18s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc3e3f4144c3445796b391010c161eab : SUCCESS in 1h 57m 05s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6cf5112d2be0458f9473c45c0da4e1b8 : TIMED_OUT in 3h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb95fe57f1fa254c686cfb3a8ef8c9642981194e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-06-22 03:08:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12090963a2ba93f648b5d439b957b08c220131cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 03:09:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"941aa9f4e2d12383226b74932ed73776f1927941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 04:40:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebfbd894b4ee4f4d9210115a926f8ed0 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fee488aebaa04773a58b2825287c1061 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4407b73f3abb47328f7165eba6058927 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58469333fc6b4465a452acbe68dc8b00 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff8e0786d21843a1a3d37c06ffc67395 : SUCCESS in 6m 50s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a2889f0c6d7c4274b2cf1a6cd85e9dbc : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2e1c4582c1e41d2743bbba72bd872a85bce07f04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 04:40:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"06988cf281c25aac2bdf13b556fcd0eff09f662d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 04:42:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfef4651ee94b708ec8182c0e52bd08 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/493613b5b02248d8a58841090539fe38 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8849cedaf70a5d76367e8f53985f536bbbe3b58e","revisions":{"8849cedaf70a5d76367e8f53985f536bbbe3b58e":{"kind":"REWORK","_number":1,"created":"2021-06-21 23:58:37.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/08/797308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/08/797308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/08/797308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/08/797308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/08/797308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/08/797308/1"}}},"commit":{"parents":[{"commit":"698c715af847b453c9595173c4601dd14ef57fb4","subject":"Fix periodic public image job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/698c715af847b453c9595173c4601dd14ef57fb4"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-20 21:03:56.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-06-21 23:58:37.000000000","tz":0},"subject":"Fix backup using customized container image registry","message":"Fix backup using customized container image registry\n\nChange-Id: Ib35743e1c315bb7fd4d0fbd99eaa409a4bf5c792\nStory: 2008996\nTask: 42658\n(cherry picked from commit 6d2ab68a8aa0406ee6320e778b8e290827c9e730)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8849cedaf70a5d76367e8f53985f536bbbe3b58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8849cedaf70a5d76367e8f53985f536bbbe3b58e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
