)]}'
{"id":"openstack%2Ftrove~799427","triplet_id":"openstack%2Ftrove~stable%2Frocky~Ie71a8776032d535248d39fa12386acda8045ad2c","project":"openstack/trove","branch":"stable/rocky","topic":"bugfix-42764","hashtags":[],"change_id":"Ie71a8776032d535248d39fa12386acda8045ad2c","subject":"Duplicate notifications are sent when deleting mysql user","status":"ABANDONED","created":"2021-07-05 02:36:57.000000000","updated":"2021-07-05 06:45:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f2a07027e628c910f8da2180465b1c701cd44bbb","_number":799427,"virtual_id_number":799427,"owner":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-05 05:53:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-07-05 05:45:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 03:20:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-05 05:45:11.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"64eaeee9575934c968a8e5237fc4535ffbab7f93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-07-05 02:36:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b59fcba569a4f96d71e9918da677137710343fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 03:20:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db4a7529915a4e56ae960440252f19ff : FAILURE in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b5bc9e0b54a475d9794dc16a7f88f9a : FAILURE in 4m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a47da8b2407849b1bcfaa133e513e50e : FAILURE in 4m 40s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/41b25ccaf8d74b79b68465f4fcbaa9d2 : RETRY_LIMIT in 4m 52s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e4e9586367ad40f1b438901f1d8f7f02 : FAILURE in 33m 15s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/c438fde39b9e4ff3b9621bbe2333dce7 : FAILURE in 32m 53s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/3da823b29f3e47f79696b9258d9e684e : FAILURE in 33m 31s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/ed090e42543c4bd7b4bd3daba09a7c6e : FAILURE in 32m 37s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/896978850f6142468688d5ec4242bec9 : FAILURE in 32m 59s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/85810281b5364fb4b39487935248e80e : FAILURE in 32m 36s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/be7c20c410974df6930e355bafdb7be8 : FAILURE in 32m 06s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/75ec2b7f776c45f3b6bac22c8ed638f1 : FAILURE in 33m 23s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/6a9719a856dc41e1a7997be16f7c99f3 : FAILURE in 32m 54s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/4e13359430be4f66a7624b1916bcf2cd : FAILURE in 42m 24s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/e3d844b759e54baca1bc7e6e8b4e1915 : FAILURE in 34m 17s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/3de4f108ac6d47219ebb7f1ddb96e1c2 : FAILURE in 3m 44s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/24d673b872ec447ca4d37098666f070f : FAILURE in 5m 08s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/e140a72936d94f669fc71e2410def63d : FAILURE in 5m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88187f065a4044748c8f7559059c684d : FAILURE in 3m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b200f11c58a34d89830c7df959d137e8 : FAILURE in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6b171e541c2a4c16ccbbc1aa706b681cae8c7c72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-07-05 05:00:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ea6483a5cc3650e78b67a640d6be80254239dd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-07-05 05:10:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2e6ebd1db273b24e2289e49d3ecb2068cc8d6bda","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-05 05:45:11.000000000","message":"Patch Set 3: Code-Review-1\n\nplease test if the issue exists on master branch, if yes, then fix on master first before backporting to stables","accounts_in_message":[],"_revision_number":3},{"id":"01af74d324705254abfdd84e35bf2c8efa077888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 05:53:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3dbaec43cf4413cb97d53cb8132deaf : FAILURE in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec38151b0864494990bd27ff9704828a : FAILURE in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93dde4af8a0a41e6b903b10b83f9595b : FAILURE in 4m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0fbe85092c8446538ffb4bb93b8ac491 : RETRY_LIMIT in 4m 27s\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/f39d84a5f6e846829576051d6562c7ca : FAILURE in 37m 32s\n- trove-scenario-mariadb-single https://zuul.opendev.org/t/openstack/build/3f1368673db0425ea5cd85bb6152845e : FAILURE in 37m 24s (non-voting)\n- trove-scenario-mariadb-multi https://zuul.opendev.org/t/openstack/build/b1efe8104dab4a5fa2e810c3c7d3899f : FAILURE in 36m 36s (non-voting)\n- trove-scenario-mysql-single https://zuul.opendev.org/t/openstack/build/179bf9e007314c04bc59bb5b3ab80dfc : FAILURE in 37m 00s\n- trove-scenario-mysql-multi https://zuul.opendev.org/t/openstack/build/466d2dd609d7486a861a5974c5c42c45 : FAILURE in 34m 47s\n- trove-scenario-postgresql-single https://zuul.opendev.org/t/openstack/build/0c83908221824c1c8db777fc839335ce : FAILURE in 35m 12s (non-voting)\n- trove-scenario-postgresql-multi https://zuul.opendev.org/t/openstack/build/6c751ee4c890499ebc58910df07f8359 : FAILURE in 35m 14s (non-voting)\n- trove-scenario-pxc-single https://zuul.opendev.org/t/openstack/build/08db33ec3c2b486a83d5518ff43c5fc2 : FAILURE in 34m 40s (non-voting)\n- trove-scenario-pxc-multi https://zuul.opendev.org/t/openstack/build/6223e7f4e4b44deab7a69676bd9d9849 : FAILURE in 34m 35s (non-voting)\n- trove-scenario-redis-single https://zuul.opendev.org/t/openstack/build/40cf56783c19443bbfffc7fc086571ba : FAILURE in 34m 51s (non-voting)\n- trove-scenario-redis-multi https://zuul.opendev.org/t/openstack/build/c460333ae4284ac19d1b9adb3c594354 : FAILURE in 37m 01s (non-voting)\n- trove-tox-fakemodetests https://zuul.opendev.org/t/openstack/build/29e8655f95814b9e868bc7a322a9a5fd : FAILURE in 4m 45s\n- trove-tox-apiexamples https://zuul.opendev.org/t/openstack/build/38b15e9827994c3eb949d729b05d44fc : FAILURE in 4m 57s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7c855e049f6c4f218cb6920a4ae6b3c2 : FAILURE in 4m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07be5a8aa65f468c995e4c376004ecc0 : FAILURE in 4m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a80e6cad88314645ad79cb1d9c2df660 : FAILURE in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"f2a07027e628c910f8da2180465b1c701cd44bbb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-07-05 06:45:45.000000000","message":"Abandoned\n\nFix it on master first","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fb385e4527531999109c6bdfb4f47e3c1684097e","revisions":{"8c896847a2500de41303312a66c0b58d2c3a5e8d":{"kind":"REWORK","_number":1,"created":"2021-07-05 02:36:57.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/27/799427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/799427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/799427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/799427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/799427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/799427/1"}}},"commit":{"parents":[{"commit":"239f9c6e4fa7d7bb9bab396f79336725a149126c","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/239f9c6e4fa7d7bb9bab396f79336725a149126c"}]}],"author":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 02:16:14.000000000","tz":480},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 02:16:14.000000000","tz":480},"subject":"Delete unnecessary StartNotification in function trove.extensions.mysql.models.Users.delete","message":"Delete unnecessary StartNotification in function trove.extensions.mysql.models.Users.delete\n\nIn the logic of deleting mysql_user,trove.extensions.mysql.models send a StartNotification which is the same as the StartNotification in trove.extensions.mysql.service\n\nFix:delete the StartNotification in function trove.extensions.mysql.models.Users.delete\n\ntask: 42764\nstory: 2009029\nChange-Id: Ie71a8776032d535248d39fa12386acda8045ad2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c896847a2500de41303312a66c0b58d2c3a5e8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/8c896847a2500de41303312a66c0b58d2c3a5e8d"}]},"branch":"refs/heads/stable/rocky"},"622118b9517cafa05855e7e3985233df0ddffe6a":{"kind":"REWORK","_number":2,"created":"2021-07-05 05:00:47.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/27/799427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/799427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/799427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/799427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/799427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/799427/2"}}},"commit":{"parents":[{"commit":"239f9c6e4fa7d7bb9bab396f79336725a149126c","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/239f9c6e4fa7d7bb9bab396f79336725a149126c"}]}],"author":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 02:16:14.000000000","tz":480},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 04:54:11.000000000","tz":480},"subject":"Duplicate notifications are sent when deleting mysql user","message":"Duplicate notifications are sent when deleting mysql user\n\nIn the logic of deleting mysql_user,trove.extensions.mysql.models send a StartNotification which is the same as the StartNotification in trove.extensions.mysql.service\n\nFix:delete the StartNotification in function trove.extensions.mysql.models.Users.delete\n\ntask: 42764\nstory: 2009029\nChange-Id: Ie71a8776032d535248d39fa12386acda8045ad2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/622118b9517cafa05855e7e3985233df0ddffe6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/622118b9517cafa05855e7e3985233df0ddffe6a"}]},"branch":"refs/heads/stable/rocky"},"fb385e4527531999109c6bdfb4f47e3c1684097e":{"kind":"REWORK","_number":3,"created":"2021-07-05 05:10:43.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/27/799427/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/27/799427/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/799427/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/799427/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/27/799427/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/27/799427/3"}}},"commit":{"parents":[{"commit":"239f9c6e4fa7d7bb9bab396f79336725a149126c","subject":"Add local bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/239f9c6e4fa7d7bb9bab396f79336725a149126c"}]}],"author":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 02:16:14.000000000","tz":480},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 05:06:51.000000000","tz":480},"subject":"Duplicate notifications are sent when deleting mysql user","message":"Duplicate notifications are sent when deleting mysql user\n\nIn the logic of deleting mysql_user,trove.extensions.mysql.models send a StartNotification which is the same as the StartNotification in trove.extensions.mysql.service\n\nFix:delete the StartNotification in function trove.extensions.mysql.models.Users.delete\n\ntask: 42764\nstory: 2009029\nChange-Id: Ie71a8776032d535248d39fa12386acda8045ad2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fb385e4527531999109c6bdfb4f47e3c1684097e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/fb385e4527531999109c6bdfb4f47e3c1684097e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
