)]}'
{"id":"openstack%2Ftrove~799437","triplet_id":"openstack%2Ftrove~master~Ic9ca04810d73b25715d3c016edf080d1b6b53417","project":"openstack/trove","branch":"master","topic":"bugfix-42764","hashtags":[],"change_id":"Ic9ca04810d73b25715d3c016edf080d1b6b53417","subject":"Duplicate notifications are sent when deleting mysql user","status":"ABANDONED","created":"2021-07-05 07:03:18.000000000","updated":"2026-09-17 17:34:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c35009f38d50c02753ce2fe44fa8db0df9e57d78","_number":799437,"virtual_id_number":799437,"owner":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-05 09:07:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":-1,"date":"2021-07-06 21:41:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 09:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 21:41:30.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"4048935ed16d1fc1cd227286688b5cdc0afefbb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-07-05 07:03:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c26521ade28faf3cfd584d704663d4c254de1e94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 09:07:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbc630eca3804eff892e483635468bf8 : FAILURE in 12m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50174c2d8f9240cb92339e0d8ad29550 : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/030feaeb65404ababf190e52b1e677f6 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1535722af31145ba8996841f7bc7dde2 : SUCCESS in 9m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3d0aa9f716341e49ed5fec20654beab : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26895b86275e4df7badcc995e4b83890 : SUCCESS in 7m 50s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/859a3efef3544fd49623b82de968c6fe : SUCCESS in 11m 29s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/dbc95b7b25cd4936a25061112921be7d : SUCCESS in 1h 01m 22s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/53a934e863624f5a836595a7d940cf24 : SUCCESS in 59m 14s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46f0c84290b1456abd4054dbd895bd77 : TIMED_OUT in 2h 03m 17s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/49b9e06c4ae5445bada81d7b6edfedd2 : FAILURE in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"263d987bce790be19c4596699e50a720ab831835","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-06 21:41:30.000000000","message":"Patch Set 1: Code-Review-1\n\n\"Duplicate notifications are sent when deleting mysql user\" this is the title of bug report, to describe what is the problem.\n\nIn the commit message, the title should be what the change does, e.g. \"Do not send duplicate notification for deleting mysql user\"","accounts_in_message":[],"_revision_number":1},{"id":"c35009f38d50c02753ce2fe44fa8db0df9e57d78","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-17 17:34:01.000000000","message":"Abandoned\n\nOutdated MR cleanup.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"439e2320757814719275a812fc569d03a3a7ecd7","revisions":{"439e2320757814719275a812fc569d03a3a7ecd7":{"kind":"REWORK","_number":1,"created":"2021-07-05 07:03:18.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/37/799437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/37/799437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/799437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/799437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/37/799437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/37/799437/1"}}},"commit":{"parents":[{"commit":"b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2","subject":"Use bridge network for db container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2"}]}],"author":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 06:57:51.000000000","tz":480},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 06:57:51.000000000","tz":480},"subject":"Duplicate notifications are sent when deleting mysql user","message":"Duplicate notifications are sent when deleting mysql user\n\nIn the logic of deleting mysql_user,trove.extensions.mysql.models send a StartNotification which is the same as the StartNotification in trove.extensions.mysql.service\n\nFix:delete the StartNotification in function trove.extensions.mysql.models.Users.delete\n\ntask: 42764\nstory: 2009029\nChange-Id: Ic9ca04810d73b25715d3c016edf080d1b6b53417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/439e2320757814719275a812fc569d03a3a7ecd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/439e2320757814719275a812fc569d03a3a7ecd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
