)]}'
{"id":"openstack%2Ftrove~799454","triplet_id":"openstack%2Ftrove~master~I433e755b562065fedd600a94f9edd94de9a1e2b9","project":"openstack/trove","branch":"master","topic":"bugfix-42767","attention_set":{},"removed_from_attention_set":{"36080":{"account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"last_update":"2026-09-18 14:39:16.000000000","reason":"\u003cGERRIT_ACCOUNT_36080\u003e replied on the change","reason_account":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},"33480":{"account":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"last_update":"2026-09-18 17:21:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I433e755b562065fedd600a94f9edd94de9a1e2b9","subject":"Fix: Validate configuration ds version on instance creation","status":"MERGED","created":"2021-07-05 10:23:39.000000000","updated":"2026-09-18 17:22:25.000000000","submitted":"2026-09-18 17:21:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799454-bugfix-42767","meta_rev_id":"e972028226bff51a7f73978e041c16024279e038","_number":799454,"virtual_id_number":799454,"owner":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 17:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":2,"date":"2026-09-18 10:19:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":1,"date":"2026-09-18 10:18:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":1,"date":"2026-09-18 10:19:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"all":[{"value":-1,"date":"2026-09-18 10:19:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"value":0,"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"}],"CC":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-05 12:29:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 23:06:25.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"CC"},{"updated":"2026-09-18 08:11:28.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"CC"},{"updated":"2026-09-18 10:18:28.000000000","updated_by":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"reviewer":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"state":"REVIEWER"},{"updated":"2026-09-18 10:19:54.000000000","updated_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"reviewer":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"state":"REVIEWER"}],"messages":[{"id":"988d5858a7c110b6ce87d40c99d4ff8dc92185a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-07-05 10:23:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7c8fc64c57c0a8ba1682a40063ee7acc741483f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-05 12:29:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf0786f00bdc4dfdb31a49af6d63129c : FAILURE in 7m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c85e6afb4324797b08f7d5d6611756b : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9cfa86e7a9e45fbb296401035a7cc12 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0a39905bfa64dd8bf10bccd653573bf : SUCCESS in 8m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15e74660ef8d4d99bb86aac921913c1e : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd56b586237940008bbff011388cb7a0 : SUCCESS in 6m 15s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d70fecf616c54af8bb73e6ab94d954be : SUCCESS in 6m 48s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d3cdb47611124a9faee7c40549eac747 : SUCCESS in 1h 32m 57s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/bde585132e08488b8a8d169792289123 : FAILURE in 33m 52s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fca9934c2b044feca43fdc063811299c : TIMED_OUT in 2h 04m 11s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0a268522558043b28c56def4762e7a4c : FAILURE in 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d3889550f096f58b14f662aa3bc90adc69f95e6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-06 23:06:25.000000000","message":"Patch Set 1:\n\nNice catch, can you also add a unit test in trove/tests/unittests/instance/test_service.py?","accounts_in_message":[],"_revision_number":1},{"id":"95c3316ce32a7c73317432243984df667cf4ba47","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-06 23:08:42.000000000","message":"Patch Set 1:\n\nand I\u0027d suggest using \"Check config group for creating instance\" as the commit message subject.","accounts_in_message":[],"_revision_number":1},{"id":"4e44dcaa0dd3cc9c83269cacf323884b28306efd","author":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"date":"2021-07-12 09:07:10.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Nice catch, can you also add a unit test in trove/tests/unittests/instance/test_service.py?\n\nCould you tell me how to run a unit test in trove/tests/unittests/instance/test_service.py?","accounts_in_message":[],"_revision_number":1},{"id":"17f50dbed65eacb231626220f53f4f8ca726cafa","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-12 21:37:50.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Nice catch, can you also add a unit test in trove/tests/unittests/instance/test_service.py?\n\u003e \n\u003e Could you tell me how to run a unit test in trove/tests/unittests/instance/test_service.py?\n\nIn the code directory, run:\n\n```\ntox -e py38 -- trove.tests.unittests.instance.test_service.TestInstanceController\n```","accounts_in_message":[],"_revision_number":1},{"id":"e9392d75c8c544b50ff1486b368c7a0cd63fe948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 06:18:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b2e7f0f4f4771b9f05d7b367d3a8bbe2f10e38d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 06:33:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ef3195b58648b8a67737db0f4a4fa2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe4b742f087b469891a7410d4040394d : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f3f597bf7e6047fc9cc7c63879575aa6 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cf9e98f560d24d96a40833b7bf523909 : SUCCESS in 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"10792f72356d43edf8a9e7469220ff0341cf4189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 07:57:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca4176067c4a4207859c35987843c858\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4dff0e3b2324b4aad6e4d85dbf6ca8c : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb1de9c39664412a5b81677c630985d : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1aef4596f20a44809681df1c6ee01da9 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b575462be7944efbb7ba201060296749 : SUCCESS in 7m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/614d05ed4777478784a1b8ca1e6618db : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d8f21d14a94753ba82c4140df16de6 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dff3e93b027f42f2a8ee17702428552f : SUCCESS in 8m 33s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/f46cb8a576b24b4c9671be732ef37265 : SUCCESS in 51m 56s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/cdeedaf476b34923b5a25390f99dc03d : FAILURE in 17m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/2e1ae902864c46d7833a62bc537fa691 : FAILURE in 25m 45s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/bc1da1617b244a7eafc2dd920ff4e72a : FAILURE in 57m 17s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/6ab23bec1f184105943c3d33f6183bca : FAILURE in 18m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/037b0b4ce390493e82db8acd1637e38b : SUCCESS in 1h 14m 18s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/5d963aa260af450983c9ce1b00fe9e08 : SUCCESS in 32m 37s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/fdccc3cd8a6d43ed941e407171408bd1 : SUCCESS in 1h 13m 34s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/7e1cf6085a1044c3b8835f5ee9edb0fb : SUCCESS in 59m 29s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/8c01f0486af0415f9628d4286f5340ef : SUCCESS in 46m 43s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/21ae8d147a364dddb1e0dc4706fa7ba2 : SUCCESS in 1h 01m 16s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/d054d7aa0d8140e18f694e045cbeb882 : SUCCESS in 1h 11m 05s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/8955e6a352bc4ae491604db6a0f92e4f : SUCCESS in 37m 52s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/003cbf3c87ce439d91117c83783d35d4 : SUCCESS in 1h 36m 04s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/f43430cc1ce54823861d2098db411e13 : SUCCESS in 39m 37s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/074ba7507c7b42b7bc24e6c0f9db8624 : SUCCESS in 51m 01s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/90600b15f6204b2ea9cbb7de23093310 : SUCCESS in 55m 08s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/799caaad51014653a5a284a2088bc764 : FAILURE in 23m 11s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/23cd4230b5bf4048a4fef80bef4e70d1 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"526e0edba3d53cfe7844995861b0fc6ffdf6c035","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 08:10:20.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"904acac34bcbbb2c3cef5c9da3c210de1ab55f99","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 08:11:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2bb74e7b827b1b53e45a39b99b6193d9075aa71f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 08:26:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3732d17d7f13468aa24ae3e729298571\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9bfb930f7bb4496787db8154e3915740 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0dbc7687e38b474fb057d3da276ed02c : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/64b2a8f04255419998cc09f8a1e52735 : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d20d5e85c826b92600454893aff108b6610b1d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 09:37:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/495b49c3a4a1434999bf693a6e6bb05b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39b78d5d76464bf4a89b7f3a1c47ad48 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ffd5db8cd84cbfa906d348ac854750 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/601e63f2c5824c8b910e408b1c7ecbec : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9553ae939b924c51a2a9a470daf181e9 : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3199759c135d48d8adc66d1fdbf69cd2 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cbf50904b4b4893bc5c9cddc769fe3f : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4db05bbf324140b580d3071ca1a176f4 : SUCCESS in 7m 53s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/740e0ce61cca4a92a3730aef5f1b2928 : SUCCESS in 37m 31s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/7164c103cb004be1acf1fb49c66139cc : SUCCESS in 1h 15m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/b3ff9673860a482a9dd8daa41559715f : SUCCESS in 1h 03m 28s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/e6351f7e098a40f1967a48f1f64de537 : SUCCESS in 54m 21s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/15217f064fe94cebb1f84be85ac2c16d : SUCCESS in 1h 05m 20s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/39a9d6ef9fb64be486c412c292bbb353 : SUCCESS in 1h 14m 08s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/2f57ea3fde404c0f90467fa0279963ff : SUCCESS in 33m 51s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/a86de77683f74d0790d0360d6231a693 : SUCCESS in 1h 24m 23s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/7e21ba820fc841c5b2813799cf2040ae : SUCCESS in 57m 57s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/d50261b367d6406e88279169ea4b73bb : SUCCESS in 49m 26s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/3be8f240214d49c4919558ceb204c494 : SUCCESS in 1h 00m 27s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/9facdba078ca4af4a19f10b9ca8cc4ce : SUCCESS in 1h 10m 04s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/85de44e609014cb4b36c362ac042ec81 : SUCCESS in 39m 12s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/b864de5cb3474315b768e99ac49a0acd : SUCCESS in 1h 09m 27s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/0624260de5374e778cf8ca03e406303e : SUCCESS in 40m 27s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/d522c6a6336749ee91256742a9f68096 : SUCCESS in 1h 00m 02s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/f2389269110e4c568e29ca9bd3d9e665 : SUCCESS in 53m 06s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/bc00842f669843798ab907fd996cebb5 : SUCCESS in 57m 20s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/45975148ec274f8c8322f3e3186b3a89 : SUCCESS in 3m 32s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/fips/enable-fips.yaml for job trove-tempest-ubuntu-ssl-replication-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/multinode/enable-multinode.yaml for job trove-tempest-ubuntu-ssl-replication-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/pre.yaml for job trove-tempest-ubuntu-ssl-replication-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/devstack-tempest.yaml for job trove-tempest-ubuntu-ssl-replication-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/post-tempest.yaml for job trove-tempest-ubuntu-ssl-replication-mariadb.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"54f353900c51f8e84208720c891b94f7d32919dd","author":{"_account_id":39300,"name":"Olessya Khussainova","display_name":"Olessya Khussainova","email":"olessya.khussainova@ps.kz","username":"Cellofun"},"date":"2026-09-18 10:18:28.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d0961f2dda6b89c814b8f14cebbfed30083d874","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 10:19:54.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cc022ba63829a8b77a262f642f1a11c4d7f218ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:20:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c3ccc2cfc1d24f8cba376fef5137bc532e4b3673","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:03:10.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/568debeae9174445bfa818530b41a9b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d03bc7edca45f9929bcbed87f6c08a : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8ba0915f87c466ca4bbaa4d9288c37b : SUCCESS in 5m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c97eb0f89a24dfca16cf48199683125 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faedfa55539242c594f45224ba441367 : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b00a417ccc0d47be88c4b4474c3ff8f9 : SUCCESS in 8m 02s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/b5b7efd7f8f447c5962be0ccf93aa157 : FAILURE in 30m 31s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/258497cb9a38435a92c292703a99f22f : SUCCESS in 34m 46s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/5e5c3c1528cf4ef7b13e00f671834919 : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":3},{"id":"7aafd66e8829e7a32651af228853926e975f6646","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 11:21:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c7b55aeeb3792c044271742f67b3f7e59b0281cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:59:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe79c84b3314f78b8b016f9894351f5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d738af9de594f18b44cb62655544c68 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f781f9bb3d054118bb1e04c812769e75 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bc6b1016e7c14b3a848a711a802b97f3 : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"63987f5f3c90198dfa787c6c7b1628e1d8920b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 13:22:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cc4efcffcbb438c993e2d67132253bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fee1eb7ca8246ac803c2139ec83bf1f : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ec31a3be7345799b7549cd7b6e0c66 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35240eaf39ae45c3b766f6ad360b02e3 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c600d148c5de425baff307564173315e : SUCCESS in 7m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/75828083e60842788faf07421e3f556f : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a4c8b78fb344fee8268101494e797a5 : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3911ff5f19084ad2958f9efc1006fe1c : SUCCESS in 8m 41s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/af655a777cae40a183f37e912a2ca844 : SUCCESS in 1h 04m 03s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/5c107494cc1c406588dfeaf1e550e1f1 : FAILURE in 22m 32s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/92362fa2bb084e7aa95df6ed9e4844e8 : FAILURE in 25m 41s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/d6a2369e37064aa285d355dd46582f82 : FAILURE in 58m 04s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/2ddc5417b12a415fbf8234d748bc9846 : FAILURE in 20m 59s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/08e5c95e7e7543a3943cb4fe2e0899b0 : FAILURE in 21m 18s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/986bdf133d86431e852951a5a0691bb4 : SUCCESS in 42m 33s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/ba4ec10de23d45b8816942fc29dd31cb : SUCCESS in 1h 38m 43s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/d593e0d4f6af4615992cda7fa000300d : SUCCESS in 1h 40m 31s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/32bcb74761534a57a371ad2292c2b0c6 : FAILURE in 53m 28s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/304b0ab736e24a039e39dc6254057731 : FAILURE in 1h 12m 44s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/36251c0e03554131b721e6bbe56b735e : SUCCESS in 1h 31m 52s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/c23a0fd6300742c5b25e8e8e23611c71 : SUCCESS in 48m 49s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/b222418205034d4eb263d6984594bf15 : FAILURE in 1h 05m 34s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/255ac9f2ecbe4bf5a01e4614b27eb904 : FAILURE in 50m 57s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/10e81d7957d741f3b7b3238648f8d3aa : SUCCESS in 1h 05m 58s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/6180c316b943495d88bf8a36fa50c575 : FAILURE in 20m 34s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/58729b86abaa468a85ea6bc237cf179d : FAILURE in 25m 29s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/98ad6deaab9a4623ad3f1adf12271881 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c61ebef34cd198297c2690f92fe4c835259c11e3","author":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"date":"2026-09-18 14:39:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7466048bb3c656683083263fe554fb29dd2e154","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 14:54:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff4e15c64dd8493aa6c075298294ef9d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/acf46c66cd364ae58fb7efc5f748674d : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fabb5b4c394448af8058526bbdc1a296 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/77381f04146d46e6a63cb93d4b95d8a7 : SUCCESS in 13m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ceef5aa38782d78b7a1933351f4a756dce299bd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:44:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/189365d876e54ae99993eaf25daed6d0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0a031b471ab4754a8389ca73956a394 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9699523df2e84d058b94c962cdc488a0 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15b1d757eb774606b373b86b06644317 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34ec8a80bdba40afae6e1d78bd0ff919 : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d7c5ed320a8d4802b9d632d9651e6b6a : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72ca7069d064c82a501cfbd90fa222b : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09309bf6bc4f4beb90d504a63c2ad237 : SUCCESS in 5m 57s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/1a153ee7000643359725f1c547155b12 : SUCCESS in 35m 47s\n- trove-tempest-ubuntu-backup-mysql https://zuul.opendev.org/t/openstack/build/25f3ff8d34924f219d5244ee9f0b40ca : SUCCESS in 1h 26m 36s (non-voting)\n- trove-tempest-ubuntu-replication-mysql https://zuul.opendev.org/t/openstack/build/6b05b42654724f6f9272be1e249d1864 : SUCCESS in 1h 06m 36s (non-voting)\n- trove-tempest-ubuntu-ssl-mysql https://zuul.opendev.org/t/openstack/build/ad659c62262f4d20a94c6f8fabe70749 : SUCCESS in 57m 33s\n- trove-tempest-ubuntu-ssl-replication-mysql https://zuul.opendev.org/t/openstack/build/b303e7bf99bc432ab8fccbed19f48c9f : SUCCESS in 1h 07m 28s (non-voting)\n- trove-tempest-cinder-storage-driver-mysql https://zuul.opendev.org/t/openstack/build/f6764478578e41eea5b711fa5f5d9d06 : FAILURE in 20m 30s (non-voting)\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/6908c3f0d64740188a4ec9583ad82fe9 : SUCCESS in 34m 39s\n- trove-tempest-ubuntu-backup-mariadb https://zuul.opendev.org/t/openstack/build/5c1c72e2d3a641e289c81c3fb37b9fdb : SUCCESS in 1h 24m 14s (non-voting)\n- trove-tempest-ubuntu-replication-mariadb https://zuul.opendev.org/t/openstack/build/97cf225974624ed48d81b05d0d382235 : FAILURE in 20m 01s (non-voting)\n- trove-tempest-ubuntu-ssl-mariadb https://zuul.opendev.org/t/openstack/build/5e3760c625954a9ea29e8e12a809882f : SUCCESS in 50m 00s\n- trove-tempest-ubuntu-ssl-replication-mariadb https://zuul.opendev.org/t/openstack/build/3030eb41ca6b430a9db96d443c31958f : SUCCESS in 1h 01m 41s (non-voting)\n- trove-tempest-cinder-storage-driver-mariadb https://zuul.opendev.org/t/openstack/build/681ed0562f404d669ad9e71b021bf35d : SUCCESS in 1h 10m 15s (non-voting)\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/7ed1123a70264562b620f1d7aa091654 : SUCCESS in 1h 00m 33s\n- trove-tempest-ubuntu-backup-postgresql https://zuul.opendev.org/t/openstack/build/5d79fd2adc434280be8c8658dbc9b722 : FAILURE in 2h 03m 25s (non-voting)\n- trove-tempest-ubuntu-ssl-postgresql https://zuul.opendev.org/t/openstack/build/88958f6eb1e940758fd9f0db72da8c42 : SUCCESS in 40m 18s\n- trove-tempest-ubuntu-ssl-replication-postgresql https://zuul.opendev.org/t/openstack/build/58079a21e1a140df8624a40288996e82 : SUCCESS in 1h 02m 05s (non-voting)\n- trove-tempest-ubuntu-replication-postgresql https://zuul.opendev.org/t/openstack/build/dc5efaed0d7b43e69a7353e0369790da : SUCCESS in 58m 40s (non-voting)\n- trove-tempest-cinder-storage-driver-postgresql https://zuul.opendev.org/t/openstack/build/d5f1c69b9f174df78c0706cfe38d0f11 : SUCCESS in 59m 37s (non-voting)\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/103c974832b042b7a0b4ea8af0d8aba8 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"6295e650eb0215511c44e1166a58257fd7aec1e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 16:45:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"83a15eeee3317b72884cd4dbeaa0d0e85996489d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:21:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb5663a42ff473cb8b8a83b5064e7cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8439e74763477fab8983ddfd1e1ff8 : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d41106e9f82842e5a82f18786e0dd9ba : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fec729de11c046f7a7486ff8156d5af4 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c965dab8c140b68169aea59b81bda8 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c3d72b7f8d048d4a9418481be9023d3 : SUCCESS in 5m 07s\n- trove-tempest-ubuntu-base-mysql https://zuul.opendev.org/t/openstack/build/b248b16b648444f3b74812a05320df71 : SUCCESS in 34m 02s\n- trove-tempest-ubuntu-base-mariadb https://zuul.opendev.org/t/openstack/build/16719e4a6b60446b803cb5d81ba24c5c : SUCCESS in 31m 53s\n- trove-tempest-ubuntu-base-postgresql https://zuul.opendev.org/t/openstack/build/884f81e28c3746e0aea8af75687d027c : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":3},{"id":"564927a87d3421a14d2b005ed47c2e04bc53c744","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:21:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e972028226bff51a7f73978e041c16024279e038","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 17:22:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/888a54cb5a65480dba2db98d74522a6e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7177c44639b4e47ac54fc93f7af0f70 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec05641e980a453fb51a7e5e71c752ce : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1a004bac30a215f381b5d2f8d1f760eea48d5580","revisions":{"110281e6408948ec48a97928e4efa3464433d4c8":{"kind":"REWORK","_number":1,"created":"2021-07-05 10:23:39.000000000","uploader":{"_account_id":33480,"name":"wocloud-trove","email":"wocloud_trove@163.com","username":"wocloud_trove"},"ref":"refs/changes/54/799454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/799454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/799454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/799454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/799454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/799454/1"}}},"commit":{"parents":[{"commit":"b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2","subject":"Use bridge network for db container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2"}]}],"author":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 08:14:35.000000000","tz":480},"committer":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 10:19:27.000000000","tz":480},"subject":"Fix:Add parameter check when creating trove instance","message":"Fix:Add parameter check when creating trove instance\n\nWhen creating a trove instance with specified configuration,add a check to judge whether Datastore Version on Configuration match the Datastore Version on the instance.\n\ntask 42767\nstory 2009031\n\nChange-Id: I433e755b562065fedd600a94f9edd94de9a1e2b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/110281e6408948ec48a97928e4efa3464433d4c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/110281e6408948ec48a97928e4efa3464433d4c8"}]},"branch":"refs/heads/master"},"687265dd4ac6f848ba3f4ca068b44530e700d798":{"kind":"REWORK","_number":2,"created":"2026-09-18 06:18:58.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/54/799454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/799454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/799454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/799454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/799454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/799454/2"}}},"commit":{"parents":[{"commit":"56999c6b8b31c6924f0e8b1be1286663bc5bbeed","subject":"Merge \"Fix database status detection for stopped containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56999c6b8b31c6924f0e8b1be1286663bc5bbeed"}]}],"author":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 08:14:35.000000000","tz":480},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 06:18:25.000000000","tz":300},"subject":"Fix: Validate configuration ds version on instance creation","message":"Fix: Validate configuration ds version on instance creation\n\nWhen creating a Trove instance with a configuration_id, validate\nthat the datastore version of the provided configuration matches\nthe datastore version of the instance.\n\ntask 42767\nstory 2009031\n\nChange-Id: I433e755b562065fedd600a94f9edd94de9a1e2b9\nSigned-off-by: wocloud-trove \u003cwocloud_trove@163.com\u003e\nCo-authored-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/687265dd4ac6f848ba3f4ca068b44530e700d798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/687265dd4ac6f848ba3f4ca068b44530e700d798"}]},"branch":"refs/heads/master"},"1a004bac30a215f381b5d2f8d1f760eea48d5580":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-18 08:10:20.000000000","uploader":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"},"ref":"refs/changes/54/799454/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/54/799454/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/799454/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/799454/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/54/799454/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/54/799454/3"}}},"commit":{"parents":[{"commit":"56999c6b8b31c6924f0e8b1be1286663bc5bbeed","subject":"Merge \"Fix database status detection for stopped containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/56999c6b8b31c6924f0e8b1be1286663bc5bbeed"}]}],"author":{"name":"wocloud_trove","email":"wocloud_trove@163.com","date":"2021-07-05 08:14:35.000000000","tz":480},"committer":{"name":"Erkin Mussurmankulov","email":"mangust404@gmail.com","date":"2026-09-18 08:10:20.000000000","tz":0},"subject":"Fix: Validate configuration ds version on instance creation","message":"Fix: Validate configuration ds version on instance creation\n\nWhen creating a Trove instance with a configuration_id, validate\nthat the datastore version of the provided configuration matches\nthe datastore version of the instance.\n\ntask 42767\nstory 2009031\n\nChange-Id: I433e755b562065fedd600a94f9edd94de9a1e2b9\nDepends-On: https://review.opendev.org/c/openstack/trove/+/1006155\nSigned-off-by: wocloud-trove \u003cwocloud_trove@163.com\u003e\nCo-authored-by: Erkin Mussurmankulov \u003cerkin.mussurmankulov@ps.kz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a004bac30a215f381b5d2f8d1f760eea48d5580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/1a004bac30a215f381b5d2f8d1f760eea48d5580"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36080,"name":"Erkin Mussurmankulov","display_name":"Eric","email":"erkin.mussurmankulov@ps.kz","username":"mongoose404","status":"PS Cloud services employee"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
