)]}'
{"id":"openstack%2Ftrove~800130","triplet_id":"openstack%2Ftrove~master~I96bb9f948dda22f50ec2a4fb5710ad9778e16b9d","project":"openstack/trove","branch":"master","topic":"iptables-persistent","hashtags":[],"change_id":"I96bb9f948dda22f50ec2a4fb5710ad9778e16b9d","subject":"Add package iptables-persistent inside guest instance","status":"MERGED","created":"2021-07-08 23:28:30.000000000","updated":"2021-07-09 13:41:28.000000000","submitted":"2021-07-09 13:40:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800130-iptables-persistent","meta_rev_id":"5392dba84170b2fc6ba8df54f3a727577628a238","_number":800130,"virtual_id_number":800130,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-09 13:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-07-09 05:21:05.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-07-09 05:21:05.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 01:37:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76f25e0c1409ef80ecd0a072dc118f778ad141f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-08 23:28:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d1958597cb1713e4e75930dbbd7172ed3cba8ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 01:37:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53229cb392704ecaa331b673b74ed437 : FAILURE in 6m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d328576cc8c7407fb35531c7d7db9429 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11c94bfc62c24422b1c151ede6618aa9 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a1f0fd6202b4514b84fc25a503f0646 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d374e120a8144d0fa286605b8a66e6d2 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ee3ff7b3ed4b95bc632035faab9818 : SUCCESS in 6m 18s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d2cf894b67a24348a975b0ba607d62d2 : SUCCESS in 6m 24s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/d701eb27eb814ad4948c49e3d8ef1cfc : SUCCESS in 1h 03m 38s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/0306e21ed3104e73bc264c6e5ff4ade6 : SUCCESS in 1h 25m 11s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3744230253bc47dfa1dd1f0348a0cacb : TIMED_OUT in 2h 03m 05s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/0400ef3c33334111b8d9dde54949636c : FAILURE in 32m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e17326f0971edf9c62ea8cfdd0148d643ae9297","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-09 05:21:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81f4ee041f221720b6951a8a95711585d152a734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 05:21:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2320d6244c95592ea3bae6581d77b9684a4f138f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 06:46:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b67726a888471e86a896ee1116bda0 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a104cb76eab4b289e4bd388b6504af7 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a203fc12c2a249d28564c18fc13fe718 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775b1069bd334ccd9d2f9b599969079d : SUCCESS in 6m 21s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/06e79e30150a45bfa18ea89d5ecf86d0 : FAILURE in 1h 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"31e94947c44f11635874c8d8d560505ef865d824","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-09 07:13:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"515f1f68fe6ea12b6d750d89f12edbad274e95ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 09:18:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bab46f12e8db485e9024cb82b5ec76f0 : FAILURE in 8m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3152775954554d5ead931d8bdae050da : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67be73d4c36b485ebc3f2a946529de15 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a0d4693d3574ec0bc2a4ea0bf869af5 : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1b0d53b44c844a8bbdad9644b335032 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ad11f90ae64cb7ac1f3bc99df3ac34 : SUCCESS in 7m 36s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/bad981fd565647d98e6204f39cf6b217 : SUCCESS in 6m 43s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/4ffe7034777d48f1a6ce7fd45934f765 : SUCCESS in 1h 23m 32s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/438cc91b0bd441589d5b085565e7394c : FAILURE in 33m 53s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95106ea9ff9e482bb004eeb6dc1bc26e : TIMED_OUT in 2h 04m 09s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/b36815c826ad4483956a87dd76cc3eb0 : FAILURE in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31e9c44dc93233b925515ee8040681b7b045d66b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-09 10:20:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba9bdf726223399d383a7cfd19249064b1515a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 12:26:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2d011fc9aeb4da19d20b9b44f190272 : FAILURE in 7m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96aab90bf8454299923b2f861a5d668a : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/092f2f4b4cd74d09b0f68e76b4478aef : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bc148cb0d2d4211822b95af7fe4126d : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01a2d56934454863a30246b6eca76bf7 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631878b969c545ecae29ee597b188cba : SUCCESS in 7m 01s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/d0087b2d76a0494cb08807690a4b3197 : SUCCESS in 7m 32s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/16729d5e26fb4973a4487a6eeb77da0d : SUCCESS in 1h 31m 39s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/80317285ec39430eb3440775a15452e5 : SUCCESS in 1h 26m 27s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ce035d2fe0d455ab3728e871a9982f7 : TIMED_OUT in 2h 04m 19s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/186daf9703864495ac4c9e6f6c6c3887 : FAILURE in 34m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88a5c96c43e4d026fcfcfd67b06f09ec95553efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 12:27:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"490d3c70b3a6a6077e023fd11f1decad259642c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:40:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e232863d0b6748fda54c92ce373c3118 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e284c0827daa4c9ca4b1f75e2036d63e : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c01de99ccd4a4a11973dd1d4bfa0d150 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52b1d17739e493a8731cf2b6eedc9a1 : SUCCESS in 7m 08s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/485fd6aa4cec4a14b8db2789a6f9b994 : SUCCESS in 1h 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a9c3d4b5d82d5c7ea8807170cb9be1c604d74cb0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:40:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5392dba84170b2fc6ba8df54f3a727577628a238","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:41:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a9f89c38b34f2d88d763184d50727c : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ce59ace17612a13bc65aea3844b4e4a187288da","revisions":{"4ce59ace17612a13bc65aea3844b4e4a187288da":{"kind":"REWORK","_number":1,"created":"2021-07-08 23:28:30.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/30/800130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/30/800130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/800130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/800130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/30/800130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/30/800130/1"}}},"commit":{"parents":[{"commit":"b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2","subject":"Use bridge network for db container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b050996b9f6df738a0f68ac36a5b5f17f8bb2bc2"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-08 21:48:24.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-08 23:27:40.000000000","tz":720},"subject":"Add package iptables-persistent inside guest instance","message":"Add package iptables-persistent inside guest instance\n\niptables-persistent is needed to persists the iptable settings inside\nguest instance.\n\nChange-Id: I96bb9f948dda22f50ec2a4fb5710ad9778e16b9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4ce59ace17612a13bc65aea3844b4e4a187288da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/4ce59ace17612a13bc65aea3844b4e4a187288da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
