)]}'
{"id":"openstack%2Ftrove~800319","triplet_id":"openstack%2Ftrove~master~I0000b9b0139471064c9589fe5c350b6facdd80dd","project":"openstack/trove","branch":"master","topic":"admin-list-instances-order","hashtags":[],"change_id":"I0000b9b0139471064c9589fe5c350b6facdd80dd","subject":"Show all the instances in descending order of creation time","status":"MERGED","created":"2021-07-10 00:44:14.000000000","updated":"2021-07-10 04:57:49.000000000","submitted":"2021-07-10 04:56:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800319-admin-list-instances-order","meta_rev_id":"41f0ec408212b336e585f07be1c5391591080b12","_number":800319,"virtual_id_number":800319,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-10 04:56:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-07-10 03:31:26.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-07-10 03:31:26.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-10 02:52:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2af87c3a190f31c17ea8dd2cc2a0fb808fbcd5ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-10 00:44:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dc1b6335e899da73d7f7900d9298132bc93e7bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 02:52:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b7df7600cc9416798d2a834b65efbde : FAILURE in 7m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4106ca7487243d9acbdd8a14c853366 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98a10061237047078ce4302f6474b1c5 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fed9925d43624b5fb2255bd1db4f306d : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3eaf1c292e684065b932f477e82e2927 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e342e05323f45909f4a5b7e6617ff47 : SUCCESS in 6m 22s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/1ff04b68326d42e98709199f07609e43 : SUCCESS in 6m 17s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/13c949f7842f4f07bce698fca446c755 : SUCCESS in 1h 02m 09s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/525a0c5146ab4a8e82791e32bfa34a25 : SUCCESS in 1h 18m 49s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39279258f1f24c5d94dfd9003d9407df : TIMED_OUT in 2h 02m 51s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/65792a3a82874954b2ef63035475becc : FAILURE in 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77163d2f72118a91f08e13371cb36a5fd3214409","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-10 03:31:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3eee907bc35a4b29911ecfb090cb382646411263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 03:31:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e676bd09d707a6f65aea836286038ab1ac5e980","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 04:56:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51cc1b238b6b4b2cad03eb49e0a1ec81 : SUCCESS in 11m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c859e621bd344d9985dcc81f4920e13 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3d7cffedb10459b894d85f67db65f14 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91374ca026344950862f1b664e1ae275 : SUCCESS in 7m 25s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/547f9dd86cc04496a5f0ed3e72a4ab0f : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7ff6386be255b18eaae70557acad1bfd1b3524d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 04:56:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"41f0ec408212b336e585f07be1c5391591080b12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-10 04:57:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64453f809b554566ab93da9e896d9360 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b4631181752be749092add04d5f0ecd57057c73","revisions":{"7b4631181752be749092add04d5f0ecd57057c73":{"kind":"REWORK","_number":1,"created":"2021-07-10 00:44:14.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/19/800319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/19/800319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/800319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/800319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/19/800319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/19/800319/1"}}},"commit":{"parents":[{"commit":"190529506b8d7bd75c9079be0c55cf9e247299a3","subject":"Merge \"Release note for adding iptables-persistent package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/190529506b8d7bd75c9079be0c55cf9e247299a3"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-10 00:41:25.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-10 00:44:05.000000000","tz":720},"subject":"Show all the instances in descending order of creation time","message":"Show all the instances in descending order of creation time\n\nThis is for list instance response of admin user, keep consistent with\nthe behavior of normal user.\n\nChange-Id: I0000b9b0139471064c9589fe5c350b6facdd80dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b4631181752be749092add04d5f0ecd57057c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b4631181752be749092add04d5f0ecd57057c73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
