)]}'
{"id":"openstack%2Ftrove~800344","triplet_id":"openstack%2Ftrove~master~I3499956a7ba0a5d0ab4a49f5938ce038446d91b7","project":"openstack/trove","branch":"master","topic":"fix-rebuild","hashtags":[],"change_id":"I3499956a7ba0a5d0ab4a49f5938ce038446d91b7","subject":"Fix rebuild instances in replication cluster","status":"MERGED","created":"2021-07-11 22:03:15.000000000","updated":"2021-07-12 11:53:05.000000000","submitted":"2021-07-12 11:51:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800344-fix-rebuild","meta_rev_id":"22d1447fb027a268c32c9543bdcd6d76dd4a424c","_number":800344,"virtual_id_number":800344,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-12 11:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-07-12 10:37:29.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-07-12 10:37:29.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 00:08:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0119f5aac1efcf2e498cd657a5e445d690901eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-11 22:03:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5442f08a135b57c70a9356fde8d8c2f7fc445f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 00:08:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d330d3cc0c14ab4a16e7d7e3fdcb178 : FAILURE in 11m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c8591fdedd4adca657c59a96d4fe59 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c56ebfb7da044565b4b3f54c19e6ba9b : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/783412cc2f754436bd7fd9ed51bd28cf : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f94aaace0e094d47bf32ca43fce7a375 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b726b88b9c3a41e99e8029632deb69fc : SUCCESS in 6m 21s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/f9f14526fd6e412db550874cd3c03829 : SUCCESS in 6m 02s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/24186885c03e41e0a9bf198ba8a59a9c : FAILURE in 28m 42s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/ecb05e873ee44f2f80827ae6a4c662d9 : SUCCESS in 1h 11m 04s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a8b39f112884fbaae07f75338192c89 : TIMED_OUT in 2h 03m 50s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/e081fedcbe5d402faf31c7275ea615c7 : FAILURE in 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a97521826c27a8f96481924c745df3dcc4adcbe9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-12 01:36:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd5be60d8146dc496066d7ea306928b2ac3efd2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 03:45:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42a3fb33bce748d4911b63f83820db74 : FAILURE in 6m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1a37ec307f48a289ede910a8fd2cc4 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1959d679d37e4decab78df4afa01665a : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3ce186e6bf04a92a87dadf5e54b3aee : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d804e470e3843cb86bd627059a79071 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e4563e1035444aa0a0fb4928e72e85 : SUCCESS in 6m 30s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0f66ab5d767043ae9218fb7660e7a5d3 : SUCCESS in 6m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5bb8bda07aed47a98f8672d06260b674 : SUCCESS in 1h 03m 58s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/83d62d5ebb8e45da9c1138cf28f73fa9 : SUCCESS in 1h 15m 17s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/254a34cbd51a406d9f3bcc9563c5a8c2 : TIMED_OUT in 2h 03m 17s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3e2e66ccafb0456280412c2281b6948f : FAILURE in 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"039ef705b495d24dbe0da8327cd8c536d50d6673","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-12 10:37:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e415dd7ffb574d991c67710ff4524db9a71585f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 10:37:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"531924f67c6db24723cd4dafeb4b4e507f14f2e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:51:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b44e97edf23b4306a525d82d353c5117 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0fcffc1edd24093ac75875819f6ceed : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2580bfeca3324f4a86620055db98f044 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1595f4fb0421480fa07867c80476cce8 : SUCCESS in 6m 37s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/edbef4a4cba64d20a67c8d75d0ebecd2 : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":1},{"id":"25f2672c495369ba668a1792349b81d2afd0f843","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:51:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"22d1447fb027a268c32c9543bdcd6d76dd4a424c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 11:53:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66d33e909f4435f892fc53c395c5159 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f0bf45ea120f5b22be50fdc0f588c431fea0e8f5","revisions":{"f0bf45ea120f5b22be50fdc0f588c431fea0e8f5":{"kind":"REWORK","_number":1,"created":"2021-07-11 22:03:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/44/800344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/44/800344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/800344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/800344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/44/800344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/44/800344/1"}}},"commit":{"parents":[{"commit":"190529506b8d7bd75c9079be0c55cf9e247299a3","subject":"Merge \"Release note for adding iptables-persistent package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/190529506b8d7bd75c9079be0c55cf9e247299a3"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-10 03:50:14.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-11 22:01:56.000000000","tz":720},"subject":"Fix rebuild instances in replication cluster","message":"Fix rebuild instances in replication cluster\n\nStory: 2009043\nTask: 42806\nChange-Id: I3499956a7ba0a5d0ab4a49f5938ce038446d91b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f0bf45ea120f5b22be50fdc0f588c431fea0e8f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/f0bf45ea120f5b22be50fdc0f588c431fea0e8f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
