)]}'
{"id":"openstack%2Ftrove~800382","triplet_id":"openstack%2Ftrove~stable%2Fwallaby~I3499956a7ba0a5d0ab4a49f5938ce038446d91b7","project":"openstack/trove","branch":"stable/wallaby","topic":"fix-rebuild-stable/wallaby","hashtags":[],"change_id":"I3499956a7ba0a5d0ab4a49f5938ce038446d91b7","subject":"Fix rebuild instances in replication cluster","status":"MERGED","created":"2021-07-12 20:04:09.000000000","updated":"2021-07-13 01:46:26.000000000","submitted":"2021-07-13 01:45:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800382-fix-rebuild-stable/wallaby","cherry_pick_of_change":800344,"cherry_pick_of_patch_set":1,"meta_rev_id":"90398a953b125bb44a2ce3dcc527dcbb8033b084","_number":800382,"virtual_id_number":800382,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-13 01:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-07-13 00:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-07-13 00:13:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 22:46:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26eb794365ca2596caaf0fe99c6ea31a47e2c92f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-12 20:04:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"507f4318e9f639dcba9d9693c5c3f8d59cb3a753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 22:46:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47ecbd9f98db4330b03330d691d782f0 : FAILURE in 7m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27bd5522d4734cca94df5e4d4fa80ed4 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a160469de04486198fdebecf9f3acf4 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3a148af7d164cb7ab7ac7d5d47655e8 : SUCCESS in 16m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/882d6e25ee864dc3970e29540413d5ac : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b8678452e040f79495a0f6dc076d29 : SUCCESS in 5m 55s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/9fc9bb31c3fd405ba10cb76e6e85ec48 : SUCCESS in 6m 05s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/5ce7c78e9ffd4d2c9b6c38d533109641 : SUCCESS in 1h 03m 37s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cef4879342234ddbb699b31bd60aeaea : TIMED_OUT in 2h 03m 35s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/c2a70f60439749b0a6a9414a5a76ebf1 : SUCCESS in 2h 40m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"791aae80a2757c2f31807c9df8a55dae8cfbd4ea","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-13 00:13:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"baa601f73846a9e6ad25f0e0d3f1ff0d02520f20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 00:13:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a7bbeed4ce609f2a1a00c85cd259a9f9874a895","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 01:45:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ded149298114f1680c09742ca8dab29 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2287444405a0471dbc4a31d0a2803e39 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/506c3e54106244b287b4e9bf62b01787 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c75cffdf6f9463883dcd5346bdef3d8 : SUCCESS in 6m 30s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/bf7e9f4c144345adaffea8e1d111b38d : SUCCESS in 1h 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e4304ad961e29836bf05975a9733a7bb6d7c2098","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 01:45:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"90398a953b125bb44a2ce3dcc527dcbb8033b084","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 01:46:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10bbe87dce644b13acff0ccd49c155f1 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67db4d40f5944aa43eeeb3275e905f48e42688e6","revisions":{"67db4d40f5944aa43eeeb3275e905f48e42688e6":{"kind":"REWORK","_number":1,"created":"2021-07-12 20:04:09.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/82/800382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/82/800382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/800382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/800382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/82/800382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/82/800382/1"}}},"commit":{"parents":[{"commit":"ae5904f92870c2726d47d693342e2f5d24c5630c","subject":"Add package iptables-persistent inside guest instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/ae5904f92870c2726d47d693342e2f5d24c5630c"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-10 03:50:14.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-12 20:04:09.000000000","tz":0},"subject":"Fix rebuild instances in replication cluster","message":"Fix rebuild instances in replication cluster\n\nStory: 2009043\nTask: 42806\nChange-Id: I3499956a7ba0a5d0ab4a49f5938ce038446d91b7\n(cherry picked from commit f0bf45ea120f5b22be50fdc0f588c431fea0e8f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/67db4d40f5944aa43eeeb3275e905f48e42688e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/67db4d40f5944aa43eeeb3275e905f48e42688e6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
