)]}'
{"id":"openstack%2Ftrove~800452","triplet_id":"openstack%2Ftrove~master~I5e5f0b99f79ba78c82cbb12482b9167ec69e50f7","project":"openstack/trove","branch":"master","topic":"admin-get-conf-instances","hashtags":[],"change_id":"I5e5f0b99f79ba78c82cbb12482b9167ec69e50f7","subject":"Allow admin user to get instances attached with any configuration","status":"MERGED","created":"2021-07-12 10:22:26.000000000","updated":"2021-07-13 09:43:34.000000000","submitted":"2021-07-13 09:42:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800452-admin-get-conf-instances","meta_rev_id":"b0711b436cf7fd6e8b50a08ba9d8ff0ca61b94bd","_number":800452,"virtual_id_number":800452,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-13 09:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-13 08:39:12.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-13 08:39:12.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 12:30:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f9aeeeffbcd0ac20dc10fe79d6e9912e4966d4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-12 10:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"851faf186d42b5b7f38f7f246bed17d2816ed0e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 12:30:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e570d6edc77c4e97a7ae8ba52898651a : FAILURE in 7m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f355728ebe4b93866c735994a935f8 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4033ddfa4a4b4f04ba290f17b04c3e3a : FAILURE in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0683cd13dff4c039a806641d4db3a0d : FAILURE in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7312257b5df42739089c0acc0a1802f : FAILURE in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9997f343d9e4c24bf1163969178fd1c : SUCCESS in 6m 25s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/0c7e5d6a4a90429296017e02720c34e8 : SUCCESS in 6m 43s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/ab77a127fa994abdbd7af579ed12c57c : SUCCESS in 1h 00m 29s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/8b247d29ef0544528d5b77e8d77559ec : FAILURE in 31m 38s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0f46b0a0c0f4ff98f4439871a567bba : TIMED_OUT in 2h 03m 05s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/ba5813bc8e0a4eecb4b98a03c6afc503 : FAILURE in 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5eff7be56a9cd16c1bb778a81e8dca1f6203ca39","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-12 20:05:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2231906f408a664418b1291ae49dee84198f8592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 21:50:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b647d5e258e34f7fa35f7f239dec3a9b : FAILURE in 7m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f07a7067b1f40cc9d7fdd25c840a88b : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/507789ed8598455aa750a2f470120c01 : FAILURE in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95d3bdfdb32d4cc9b563b364d8a2a44f : FAILURE in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cab65f26ce024407858a10b264d0b4b5 : FAILURE in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c03e9630af42428b483a6216289b1b : SUCCESS in 7m 04s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/ac1a18b57026452297ff52bed4d7ad17 : SUCCESS in 6m 36s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/aa76b8d945684baabfad3a35e92957d6 : FAILURE in 1h 06m 23s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/07a67f154405449796ca4df8ce224443 : SUCCESS in 1h 26m 40s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb0131a7db7e400b89cdbf3e4bcd53cd : FAILURE in 1h 44m 59s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/6fd9999cd2f84e43ae0e1c1fb4090f0b : FAILURE in 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2437001a5b017b09d6420d5a9bf7fb6fc519db7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-12 21:57:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b83b706f3bbf64c3c63f3d184a2857dc406b902a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 23:56:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bbe613860e34cc6a55002e1d864f0f8 : FAILURE in 9m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b70b4aa53b6450c8fb256279622f579 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e36aa7ba617f44af944186560751b1f3 : FAILURE in 9m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27d8a208c2394a9da27f7ff8ffffa66f : FAILURE in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/093dafe07af94486be257702166af304 : FAILURE in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5107410e300457a90f9c835139658b2 : SUCCESS in 6m 40s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/b359157e33a049ed9a4b16905cc29e33 : SUCCESS in 6m 14s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/26fbdc6f4baf42f6b8260e0ff363494d : SUCCESS in 1h 01m 56s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/863f5598949f45bf92db8d5c36684254 : SUCCESS in 59m 18s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d8f31b887534bcb8df114ac496a51fe : SUCCESS in 1h 55m 06s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/224970fde5ce49ecb23d42dd223338d0 : FAILURE in 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8920a768f0d00b8f751f4b7f4a854f08539174a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-13 00:21:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"451c586e2e98ef9e4c162f3197890cda0707de73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 02:14:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e49dd53ac954174abc49301b12acbf7 : FAILURE in 7m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f63d6ead5a474689168f99e040ab66 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/135f65efe8064a0c8350a084def727e5 : FAILURE in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecbef8495e6c4a249bb41c858008ff78 : FAILURE in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d00d2f1e8c9c4a998502fefc182adb80 : FAILURE in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed90f4dac95c4803bef8551249d5611e : SUCCESS in 6m 35s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/7846243b92b4467b8c8605be39f12dbf : SUCCESS in 6m 24s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/96d1ad8092cc4705b79b0e23d69dd2da : FAILURE in 32m 02s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/21c98fed066a4bfcb672b925ef4ad676 : SUCCESS in 1h 22m 41s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f29b53336ac14da4bc28a7da9829dd66 : SUCCESS in 1h 48m 59s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3c89cea8e0d94e00bb40addf9830b26f : FAILURE in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b15a57c00d4b1f58c83987269b4e5e489b96331","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-13 04:05:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a01f5a5ffcf92655e203d7f28490678906f639d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 05:54:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c492eab16c44217b4729074fe8b9cd6 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331f20c24a324f1f8a4f81a7be7ccd20 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6bd129e4a114021a0e716b45fd13238 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7577563b2da43d48e814d96ac853919 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5447994a3e4c4864a3a77eccba43910d : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1944b6a11045faa6978c6b4297f833 : SUCCESS in 6m 13s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/c9a2ce1382b240c5972094bb659f4bcc : SUCCESS in 6m 05s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/3d5751ce1f0e45658b657e513f56c749 : SUCCESS in 1h 25m 02s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/cacb9f51c63c41fcab41660070f85b34 : SUCCESS in 1h 02m 26s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37203db0afd04f3d9f2d4caed1233575 : SUCCESS in 1h 45m 17s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/3732494f4b8348c38ebdad1e7fd2a9a5 : FAILURE in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"80b985721b80f11d4cd02efb30af708d54ff263a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-13 08:39:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9ee2c6ae9015715b314e804f0c2da1fb735020f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 08:39:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b99936f9133850e157fa7e86d5c52b2b727fdfb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 09:41:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c166d0b10aa4a8495cecfb874cbec86 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/445b2086875f41b481bb7918a6ae7a38 : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8db09b6393804e048e94e87f633e4671 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4d8bba53544484aa96727bc974d0db : SUCCESS in 10m 52s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/e98412ab96634bf2a99d4618f9226d90 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":3},{"id":"661227f8a788cb350168133d69057ab26c0e61fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 09:42:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b0711b436cf7fd6e8b50a08ba9d8ff0ca61b94bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 09:43:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73d1598697247e4bb2e45d4989c7a0f : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"24fc4a53dcf1140349e0ae32386046ff597e416b","revisions":{"b9f81263d49c7f9967885eaefefc73d21969a357":{"kind":"REWORK","_number":1,"created":"2021-07-12 10:22:26.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/52/800452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/800452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/800452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/800452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/800452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/800452/1"}}},"commit":{"parents":[{"commit":"7b4631181752be749092add04d5f0ecd57057c73","subject":"Show all the instances in descending order of creation time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b4631181752be749092add04d5f0ecd57057c73"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-11 22:41:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-12 10:21:37.000000000","tz":720},"subject":"Allow admin user to get instances attached with any configuration","message":"Allow admin user to get instances attached with any configuration\n\nChange-Id: I5e5f0b99f79ba78c82cbb12482b9167ec69e50f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9f81263d49c7f9967885eaefefc73d21969a357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b9f81263d49c7f9967885eaefefc73d21969a357"}]},"branch":"refs/heads/master"},"b550e8cdd7a8054ef549e763d053c42fe2ab6544":{"kind":"REWORK","_number":2,"created":"2021-07-13 00:21:03.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/52/800452/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/800452/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/800452/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/800452/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/800452/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/800452/2"}}},"commit":{"parents":[{"commit":"7b4631181752be749092add04d5f0ecd57057c73","subject":"Show all the instances in descending order of creation time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b4631181752be749092add04d5f0ecd57057c73"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-11 22:41:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-13 00:18:26.000000000","tz":720},"subject":"Allow admin user to get instances attached with any configuration","message":"Allow admin user to get instances attached with any configuration\n\nChange-Id: I5e5f0b99f79ba78c82cbb12482b9167ec69e50f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b550e8cdd7a8054ef549e763d053c42fe2ab6544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/b550e8cdd7a8054ef549e763d053c42fe2ab6544"}]},"branch":"refs/heads/master"},"24fc4a53dcf1140349e0ae32386046ff597e416b":{"kind":"REWORK","_number":3,"created":"2021-07-13 04:05:42.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/52/800452/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/52/800452/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/800452/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/800452/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/52/800452/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/52/800452/3"}}},"commit":{"parents":[{"commit":"7b4631181752be749092add04d5f0ecd57057c73","subject":"Show all the instances in descending order of creation time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/7b4631181752be749092add04d5f0ecd57057c73"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-11 22:41:21.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-13 04:04:14.000000000","tz":720},"subject":"Allow admin user to get instances attached with any configuration","message":"Allow admin user to get instances attached with any configuration\n\nChange-Id: I5e5f0b99f79ba78c82cbb12482b9167ec69e50f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24fc4a53dcf1140349e0ae32386046ff597e416b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/24fc4a53dcf1140349e0ae32386046ff597e416b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
