)]}'
{"id":"openstack%2Ftrove~800622","triplet_id":"openstack%2Ftrove~master~If2d3022a59ba535e1361a3ba9bc9dd983ca25256","project":"openstack/trove","branch":"master","topic":"fix-get-config","hashtags":[],"change_id":"If2d3022a59ba535e1361a3ba9bc9dd983ca25256","subject":"Fix getting config option value for database","status":"MERGED","created":"2021-07-13 11:24:37.000000000","updated":"2021-07-14 22:50:40.000000000","submitted":"2021-07-14 22:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800622-fix-get-config","meta_rev_id":"e5cc133c4b74f3130f9d43f49fe566709dd32767","_number":800622,"virtual_id_number":800622,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-14 22:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-14 21:23:45.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-14 21:23:45.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-13 13:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"238056ebe65b922d072fac5bb109c5c5720bfaa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-13 11:24:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dc056bbdd6a741bad40ca8a45520804858ef900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 13:30:04.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02897104bee543f29fee5d2d36d362a7 : FAILURE in 10m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a2e69ab760542c3bd0e05e96f5e43c7 : FAILURE in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79d2867965c54a6eb0b9919a4601a1ec : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd467d8fa0cc4570b18db944acc49b28 : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/706c6cd70fc547d99d20a229fd0aa35f : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312ea05787ea44d98534bce50ae17fd8 : SUCCESS in 6m 27s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/484a9d39ec774332a41aaa4e4723a7a5 : SUCCESS in 6m 38s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/04c78a4710694742a1210eff23933f52 : SUCCESS in 1h 29m 56s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/e27235262b2a4d23a00ea6de5a78e400 : SUCCESS in 1h 23m 23s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0fbfa32f5114f50844d2b467d438322 : TIMED_OUT in 2h 03m 20s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/209aa2ffa7f145a6b5fbc6b19a6c9e74 : FAILURE in 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1a601207861a98e7eb8cc2d83cf4efec9ede895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-14 08:52:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c18a8d1355b297e199509fc12be4f3f769f1ab5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 10:56:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73cea6f6978c4c74bfeb1cd0c14728f6 : FAILURE in 7m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f612aa4cbb49109a0b3a94c7532dc9 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10eba85fb78941e8bbcd30c4e95c4a02 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf66301c6a2c45dfa0dae6af01d7a0dc : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ccbf9f1c0d34bbeafe146a794a60d98 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e14ef93723047c282c34a8500fb1265 : SUCCESS in 6m 29s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/168a86be01034a10957181fecd9cd0e9 : SUCCESS in 6m 16s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/082fb61b3dba4ed0baf673a35d2b6401 : SUCCESS in 1h 31m 05s\n- trove-tempest-postgres https://zuul.opendev.org/t/openstack/build/1f44ed6c410f462595359b7353edbb9e : SUCCESS in 1h 04m 43s\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11bff297befa41d8a27df57a558e7748 : TIMED_OUT in 2h 03m 36s (non-voting)\n- trove-functional-mysql https://zuul.opendev.org/t/openstack/build/753096c400584127bab0f0119cc5b9c7 : FAILURE in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1bb2a43e124e96534e769345891f41eae4ea0ade","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-14 21:23:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f93dc086e91b59d880ec659b4236af543ac6b7e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 21:24:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad66ea6ad8d5fcb7fe7db4cf04920fdf3952e832","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 22:49:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded8696963be49589c88aacaa97337cf : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f127ddeaffb47f0a8971c8e7b958a79 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e9a7419dcde4d838390d1d6956cb971 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6459e7c631d43dd89f6b54ce7d8a1f0 : SUCCESS in 6m 56s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/a75d9b3fef89468b90c6e08fac168588 : SUCCESS in 1h 23m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8e9ad2208833174ba3222f4ff5bbe59884d50b06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 22:49:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e5cc133c4b74f3130f9d43f49fe566709dd32767","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 22:50:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d8cc94ff8f458b96ed3e4ffc09cf66 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69f08ab470a0d1d1d4676b41ad29a9c19ce28648","revisions":{"c217d9ca3f528602a997328fe4df6234d96da6f8":{"kind":"REWORK","_number":1,"created":"2021-07-13 11:24:37.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/22/800622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/22/800622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/800622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/800622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/800622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/22/800622/1"}}},"commit":{"parents":[{"commit":"437d594e86b3efd9a797ab4462681404b31b41e8","subject":"Merge \"Allow admin user to get instances attached with any configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/437d594e86b3efd9a797ab4462681404b31b41e8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-06 21:34:17.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-13 11:03:42.000000000","tz":720},"subject":"Fix getting config option value for database","message":"Fix getting config option value for database\n\nChange-Id: If2d3022a59ba535e1361a3ba9bc9dd983ca25256\nStory: 2009051\nTask: 42821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c217d9ca3f528602a997328fe4df6234d96da6f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/c217d9ca3f528602a997328fe4df6234d96da6f8"}]},"branch":"refs/heads/master"},"69f08ab470a0d1d1d4676b41ad29a9c19ce28648":{"kind":"REWORK","_number":2,"created":"2021-07-14 08:52:37.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/22/800622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/22/800622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/800622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/800622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/22/800622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/22/800622/2"}}},"commit":{"parents":[{"commit":"437d594e86b3efd9a797ab4462681404b31b41e8","subject":"Merge \"Allow admin user to get instances attached with any configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/437d594e86b3efd9a797ab4462681404b31b41e8"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-06 21:34:17.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-14 08:52:28.000000000","tz":720},"subject":"Fix getting config option value for database","message":"Fix getting config option value for database\n\nChange-Id: If2d3022a59ba535e1361a3ba9bc9dd983ca25256\nStory: 2009051\nTask: 42821\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69f08ab470a0d1d1d4676b41ad29a9c19ce28648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/69f08ab470a0d1d1d4676b41ad29a9c19ce28648"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
