)]}'
{"id":"openstack%2Ftrove~800857","triplet_id":"openstack%2Ftrove~stable%2Fvictoria~If2d3022a59ba535e1361a3ba9bc9dd983ca25256","project":"openstack/trove","branch":"stable/victoria","topic":"fix-get-config-stable/victoria","hashtags":[],"change_id":"If2d3022a59ba535e1361a3ba9bc9dd983ca25256","subject":"Fix getting config option value for database","status":"MERGED","created":"2021-07-14 23:38:53.000000000","updated":"2021-07-15 06:25:13.000000000","submitted":"2021-07-15 06:22:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800857-fix-get-config-stable/victoria","cherry_pick_of_change":800622,"cherry_pick_of_patch_set":2,"meta_rev_id":"49fe0fbd3fd886424c8584467087ff5e89c8b0b3","_number":800857,"virtual_id_number":800857,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 06:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2021-07-15 04:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2021-07-15 04:31:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 01:58:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43164e2f7aadcb6b46966974101b060a1f849787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-14 23:38:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c155890ae0e52bc63ecb2ba399a6ab48028c1cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 01:58:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/012761f504a54e20aba0fa37c30afcd0 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3af5b48cf1e4a2b8bd5cb00c289e2e5 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f979d09fa34d4a3484289cf255bf0f90 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e848f6446b84e2ea5b4f155f79041d1 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bdcc9768e60422bb75c400622170055 : SUCCESS in 6m 03s\n- trove-tox-bandit-baseline https://zuul.opendev.org/t/openstack/build/870e95b000d44c12af73d17d9476f67a : SUCCESS in 6m 41s (non-voting)\n- trove-tempest https://zuul.opendev.org/t/openstack/build/48b8c875889645a882621d730f1d2497 : TIMED_OUT in 2h 13m 21s (non-voting)\n- trove-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9e0f91cf5d04180a4651b574ca5fe0f : SUCCESS in 1h 57m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4862b0fac41e848d80e5a3879bc8f31b0f2cef6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2021-07-15 04:31:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd005ed1cc5f866955c7f612f7bb4b4448b09997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 04:32:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d4bba24f9728a197c4296b1c98b2adb10f2121a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 06:22:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b6c29213ab4e3da7cd5dcb4b1a4ee9 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1140dbdb08448a393f2ae1b200c6e79 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d562bea5e5248df8552f338a2800f81 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb373d187f0343ba80cd8f20443016da : SUCCESS in 6m 34s\n- trove-tempest https://zuul.opendev.org/t/openstack/build/84da5252cca442fd8ce77eb215acbd7d : FAILURE in 1h 50m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5cde71d0640d21114d97b0520302799d7bd6754","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 06:22:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"49fe0fbd3fd886424c8584467087ff5e89c8b0b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 06:25:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db33845476314990b0b412a15e4655c4 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be65d641d28fa84dc0629ecc29b6799eb99fa195","revisions":{"be65d641d28fa84dc0629ecc29b6799eb99fa195":{"kind":"REWORK","_number":1,"created":"2021-07-14 23:38:53.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/57/800857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/trove","ref":"refs/changes/57/800857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/800857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/800857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/trove refs/changes/57/800857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/trove refs/changes/57/800857/1"}}},"commit":{"parents":[{"commit":"327231dcc7887639202ba60d152d81bc9221b17a","subject":"Keep user defined configuration after resizing instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/327231dcc7887639202ba60d152d81bc9221b17a"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-06 21:34:17.000000000","tz":720},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2021-07-14 23:38:53.000000000","tz":0},"subject":"Fix getting config option value for database","message":"Fix getting config option value for database\n\nChange-Id: If2d3022a59ba535e1361a3ba9bc9dd983ca25256\nStory: 2009051\nTask: 42821\n(cherry picked from commit 69f08ab470a0d1d1d4676b41ad29a9c19ce28648)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be65d641d28fa84dc0629ecc29b6799eb99fa195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/trove/commit/be65d641d28fa84dc0629ecc29b6799eb99fa195"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
